)]}'
{"id":"openstack%2Fblazar~952471","triplet_id":"openstack%2Fblazar~master~I5a62b7b6ffdefafa1966a74faa2725ff88fdec3a","project":"openstack/blazar","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-26 15:16:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2025-06-26 15:12:56.000000000","reason":"\u003cGERRIT_ACCOUNT_33582\u003e replied on the change","reason_account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-23 08:43:00.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I5a62b7b6ffdefafa1966a74faa2725ff88fdec3a","subject":"Remove support for Python 3.8 and 3.9","status":"MERGED","created":"2025-06-12 15:17:09.000000000","updated":"2025-06-26 15:50:35.000000000","submitted":"2025-06-26 15:49:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952471","meta_rev_id":"5ae5ada962ba451f64498746b06cec70b7f48034","_number":952471,"virtual_id_number":952471,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-26 15:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-06-26 15:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":2,"date":"2025-06-23 08:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-26 15:16:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-06-26 15:14:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 15:41:44.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-06-12 15:41:44.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"},{"updated":"2025-06-12 16:00:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdeb19b9763dd8de71206e7c8f6f2e8b3921d3a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 15:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a68c174a43f05dfd4e040b531b7c7c5d0e9aa36a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-12 15:41:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92caef668f22a672e57a1949d076f41c282e6d20","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-12 15:44:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fbb96461f6ffa7e53b9f2abe7eb0fe52a1c0644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 16:00:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f98785766984496f852b429e53443ac0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b3c066e076a4ea5bae218b5b0220779 : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9e92a03016e49dc9dcdf56eccbf8b5d : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c4ad01ef7a049c98d3473f8a3ed42c1 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cb239d1ed5145de8ae82e73ee606091 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20768da046f3499d9a075181beab813e : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a92658fd62f4431916b086729a7a9eb : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a00cd98ec10b4611b99757641a3f2d77 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5afc709236fd48d49687c7a1cbb82d6e : SUCCESS in 3m 04s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/7ffa84d751ea4eecaa1ae212c8dd86e5 : SUCCESS in 24m 46s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/2b90b7c92c0445b2b3fb2897febaf003 : SUCCESS in 36m 54s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e26b7ddb5d1c446caa8e2848d6218bb1 : FAILURE in 15m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3cef76fe8be415fbc4df039bccc3735 : FAILURE in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45f2d00f8e0d0eb1d13576eb3535f879e6052579","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 16:38:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ba0cfb647cc3bce2019302085d6e84020117256d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-12 16:41:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c93d3ee5483d6036d92f82a4495d283cc9adaf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 17:19:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726cddbc8fc74f7e801b2ca274a977e6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c806fda28fbf4d709001d9cff534c464 : SUCCESS in 2m 15s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/251b4df58185438ab3410f23b9761e26 : SUCCESS in 3m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50285536e7f142f9ab1a16dcb0069707 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c936839becf64000a3b6d0e96043c11b : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46690594bbab4909bac89eded863ba30 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ff8ed3bcdf841ad8a4d1c5c44db4e74 : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/94b76957453d4b04bf225f0f9c2ba2b2 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01decaa8e9fb495482f5716f05d91821 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e71fb36ca9ed47d18f8c77a4c5285c4c : SUCCESS in 3m 58s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1ea6d11aea3d4b46bfa4853f61a356fc : SUCCESS in 35m 32s\n- blazar-tempest-plugin-jammy https://zuul.opendev.org/t/openstack/build/cfd820131ae047c38805749e026fba59 : SUCCESS in 34m 35s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/54c3bdfd69d6410ca481bce3299872e5 : SUCCESS in 39m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ba565df92e2345cc8236a5d07764354e : FAILURE in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab54601b5e7f384645cef5154c3439fa9c3c7edc","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-23 08:43:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4404f613ea2d60e01ab7bfd5e285ec5a3f9fca66","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-06-26 15:12:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c27a56d47b865aa65a86a9f889851049d8982682","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-06-26 15:14:43.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c71e82de54dd0f1e9bd37cb85a6eb7d27c1a9513","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 15:15:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9354a40e8fc537d43a77e18ab3ab3d7487deafdb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-26 15:16:04.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"369b4c6510b8555a0b6bc62c738537f168960a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 15:49:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d209c55735b64f6f85241b15c58fb13b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/87c468d6f18f4d138c1cd3c8cbf3e69c : SUCCESS in 3m 18s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3b70dbc0dc4c4a33a9b15ba365db2671 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a4ccc254464df99d21989d29c4a64e : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92a1fe77fec645799399f98281a17492 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14bca531f1b34b1c9a56b60bf3f534f0 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5e002d8cec4349bd1891d711fa9fca : SUCCESS in 10m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd999e00d3b94aeba3fa047d1f5d2457 : SUCCESS in 2m 50s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/66ffd3d8d0244bad9236b1e659abe52d : SUCCESS in 20m 34s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/da19fa15e9fd4c5a9174501f36aac6ff : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":2},{"id":"34601487ea7f6ac81804e0f69271ac92bcafdbb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 15:49:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5ae5ada962ba451f64498746b06cec70b7f48034","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 15:50:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2d8b16e729947cf97a8ca93dde288d5\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/de9b933b71d049158c84068951777ff2 : SUCCESS in 42s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbad86bd6b3c482ea1f9c2e7073601f0 : SUCCESS in 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1900fa562ae848469f181fccbd8a811f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3b235baf312fe32ccba9e931c0750afd91b59eb7","revisions":{"0e69be6635a89237beaa41c6bd4b8f156e574148":{"kind":"REWORK","_number":1,"created":"2025-06-12 15:17:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/952471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/71/952471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/952471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/952471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/952471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/71/952471/1"}}},"commit":{"parents":[{"commit":"dc106c745dc4adf893ce86a130b5c9119eb35d47","subject":"Merge \"Fix lease date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/dc106c745dc4adf893ce86a130b5c9119eb35d47"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-12 15:16:28.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-12 15:16:28.000000000","tz":120},"subject":"Remove support for Python 3.8 and 3.9","message":"Remove support for Python 3.8 and 3.9\n\nPython 3.9 is no longer part of the tested runtimes [1]. Python 3.8\ntesting was removed in the 2024.2 release [2].\n\nAdd Python 3.12 to classifiers.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n[2] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I5a62b7b6ffdefafa1966a74faa2725ff88fdec3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0e69be6635a89237beaa41c6bd4b8f156e574148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0e69be6635a89237beaa41c6bd4b8f156e574148"}]},"branch":"refs/heads/master"},"3b235baf312fe32ccba9e931c0750afd91b59eb7":{"kind":"REWORK","_number":2,"created":"2025-06-12 16:38:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/952471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/71/952471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/952471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/952471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/71/952471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/71/952471/2"}}},"commit":{"parents":[{"commit":"dc106c745dc4adf893ce86a130b5c9119eb35d47","subject":"Merge \"Fix lease date\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/dc106c745dc4adf893ce86a130b5c9119eb35d47"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-12 16:38:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-06-12 16:38:23.000000000","tz":120},"subject":"Remove support for Python 3.8 and 3.9","message":"Remove support for Python 3.8 and 3.9\n\nPython 3.9 is no longer part of the tested runtimes [1]. Python 3.8\ntesting was removed in the 2024.2 release [2].\n\nAdd Python 3.12 to classifiers.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n[2] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I5a62b7b6ffdefafa1966a74faa2725ff88fdec3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3b235baf312fe32ccba9e931c0750afd91b59eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3b235baf312fe32ccba9e931c0750afd91b59eb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
