)]}'
{"id":"openstack%2Fblazar~952504","triplet_id":"openstack%2Fblazar~master~I5f6de2316e42685a4f80aed990327ade61ccb50a","project":"openstack/blazar","branch":"master","topic":"jammy-removal","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-06-25 18:11:14.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-16 07:03:41.000000000","reason":"removed on reply"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-06-25 16:04:44.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I5f6de2316e42685a4f80aed990327ade61ccb50a","subject":"Remove Jammy job","status":"MERGED","created":"2025-06-12 19:24:18.000000000","updated":"2025-06-25 18:12:39.000000000","submitted":"2025-06-25 18:11:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952504-jammy-removal","meta_rev_id":"f2f834503d55aa4973822cb6695b1a8256e3954a","_number":952504,"virtual_id_number":952504,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-25 18:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-16 07:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-06-14 16:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-06-25 16:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2025-06-25 16:04:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 20:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-14 16:06:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-06-14 16:06:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-06-14 16:06:36.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"1eefa3ed51bbeca7dafc7b9fcfcb168531771757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-06-12 19:24:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07a5899f5148f1e6f1bc56cf2e9b04d1d4fd8bde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 20:07:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4ce8e55dbb8497a8934cbec37995d44\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c084564e0a1430d8f460e348565906e : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b9704fa32044fe9bac8a2cf7caf92e : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ecf6a314a864bf19ac728268b453cb6 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3d7cc5e78834e0dac646055facaf7aa : SUCCESS in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/961d99b94d3f4bfeb00be7366ec7ab6e : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/123e127a942b48e1b11c2ce95d782ae1 : SUCCESS in 9m 20s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/cccac0c3b4044bf681397fb0296633a3 : SUCCESS in 38m 27s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f7ff197974684fa68652db17b4e83366 : SUCCESS in 21m 02s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e77092a55cbe46f2bf4745abc5819501 : FAILURE in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5459ae5b4b63f635571d27766d8268a09b3564e4","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-14 16:06:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d49c6fe447deeb5216f0eb1d07be198de22ab4b9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-16 07:03:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d1d0ce0cebf7594db2055e504d3a8e74b6781686","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-06-25 16:04:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a5d5aa0739bbe62b82e3ad0de737ce560d25f56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:05:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fdebd632f39eda22dafe06c071e8fce14d6968c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:32:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ae2f4b7846d46ab88987186ce0e84c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/763cf0ff26454d9f9732848a2d2016e6 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aca2a4f1c17d45bebebaf01df95f5f47 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3a9f66849fd468ca383c35d0bb84c30 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7e2bc85fe64791a3b65df38c41895b : SUCCESS in 8m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4a7d39ef7b7047f1a88620207bf0995f : FAILURE in 7m 33s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/7233b2f9656645c181db9d7aba8962f5 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6007222320eec9541f7605c1d90da561022c5ced","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-25 16:32:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7a48fa85cf48add68ff3398cf51e434c1f7daf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 17:24:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e7d0f6878824f8693b7a4715a571b2f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be865dadb45498b862bcada0d48730e : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a23884ac639c420385706df04c522d04 : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/780b8692956049a4942a68fbde972a5e : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/625df85182514606a00c080f85da9123 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47e40ecbaa064870aae81f03ce601286 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a1287c499349498af936f43ffa4fd7 : SUCCESS in 8m 42s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/2f7122b75d55435ca031d3c1a69efcad : SUCCESS in 37m 17s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/da680a20f7ed4aa7a381a6c9f32143fc : SUCCESS in 21m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c0e6e0b18a684e85a420d7acda545373 : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae26508018cf379884447076275c8c106d2489b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 17:25:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00d89bd3e7e86d21e874b60d0885c3e740a51f9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 18:11:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdd3ee0ad2ef495ea9d110b6b756fe3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb549894ee248c6b7a8e1e0f77e04fe : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05112ff9d02542658aa577143b0d5256 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58c3ece5e0c146c18a5f1493b3d467a5 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e3d4ca2fce34a6b8aaefe2bbc5ebd1c : SUCCESS in 11m 19s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/9f516a71968743cc9f09efe466818cec : SUCCESS in 39m 21s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/5a8d90c1f01b4becaa65f64a03cbf083 : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":1},{"id":"88c12875bca883cf0e23a7531f07e4321fe9d457","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 18:11:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f2f834503d55aa4973822cb6695b1a8256e3954a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 18:12:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f10d44ad3e4e32b03f8d6c9c1cf876\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654c9d9c29674c8c941886ef741c0889 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"db0680a4c8dd5fa80e0ec90eccf03354161586f5","revisions":{"db0680a4c8dd5fa80e0ec90eccf03354161586f5":{"kind":"REWORK","_number":1,"created":"2025-06-12 19:24:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/952504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/04/952504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/952504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/952504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/04/952504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/04/952504/1"}}},"commit":{"parents":[{"commit":"d91fac698439241be8eb5bb290ad40aa20fe7d8d","subject":"Merge \"Support Nova ephemeral disks in flavor-based reservations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d91fac698439241be8eb5bb290ad40aa20fe7d8d"}]}],"author":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-06-12 19:23:54.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmaan@ghanshyammann.com","date":"2025-06-12 19:23:54.000000000","tz":-420},"subject":"Remove Jammy job","message":"Remove Jammy job\n\nJammy testing is not needed in 2025.2 release.\n\nChange-Id: I5f6de2316e42685a4f80aed990327ade61ccb50a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/db0680a4c8dd5fa80e0ec90eccf03354161586f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/db0680a4c8dd5fa80e0ec90eccf03354161586f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
