)]}'
{"id":"openstack%2Fblazar~953370","triplet_id":"openstack%2Fblazar~master~Ie029a034db0c5448fb3e0388763dfcd3f4b4469c","project":"openstack/blazar","branch":"master","topic":"remove-py39","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-26 05:56:35.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_9816\u003e by clicking the attention icon","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ie029a034db0c5448fb3e0388763dfcd3f4b4469c","subject":"Update supported python versions","status":"ABANDONED","created":"2025-06-26 01:22:19.000000000","updated":"2025-06-26 06:27:59.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6152e1848f01c2ff8ee65725951cd21f888ddf12","_number":953370,"virtual_id_number":953370,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-06-26 03:02:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-26 01:54:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-26 05:48:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"28db01deb7bb6409623338adf334557d841e0338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-26 01:22:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e742635998976bc489148f461f118c187fc594a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-26 01:22:52.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e102f648aed738c400459b7bf8d306d13aab2375","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-26 01:23:21.000000000","message":"Topic set to remove-py39","accounts_in_message":[],"_revision_number":2},{"id":"f3e1c4eccffa428f2cd3716e9adef6f2cf69edce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 01:54:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00354f54ff944dedaf75ba73ef47f7ea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c1f6866e08148faa9d01d66f9dbee3a : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9529fa8c4ede45d3a0505122888ad787 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01dfd3113cbf460dab0b7be61d673c51 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02ef433c0eae48b18bb20c1f1440d216 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a94ef47d4f1d42fdb2dc4c31b0e42004 : SUCCESS in 6m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63e341937eb04be383303738f3ca53e1 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde68586aaf14b6fb3f0a43bc7c383f9 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/293042efbc99489c9950d44558664ff6 : SUCCESS in 3m 13s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/a1ed944a3cbd431ba7755a835c93d332 : SUCCESS in 22m 22s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/fe6a63d55a72490486b3e9b0582a3295 : SUCCESS in 30m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fefb5eace4d34543a5f9ccb70778098a : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e81a98c85892979dbfdccd31891c1675e73af09a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-26 02:13:36.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a3c00151a1c69510f6e7a54f7e8e2a63758e4a45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 03:02:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6504d6c76f64cb4b026af8c3469fadb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b196fb6d38524aebbe1cc841546968dd : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd1eb2a6f17a407db1e3b4cd3ccede2c : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37fff8897aa431a88f1b39fc91cfbf2 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b736df736dd94efcbde1f128d554ab8a : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20ead05be34445a6bab92087c9ddb618 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72624a3c3be84ff0a21d62ec8ac75ef6 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f6786da08e437784bcf9c33d7c84dc : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e21be98060a418a8a8e3e5e5026e7bf : SUCCESS in 3m 07s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/40154c05b4c546cfa0ec485732dce802 : SUCCESS in 42m 05s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e44765867afb4ce1abca3efcbc417d70 : SUCCESS in 38m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/73292cd2bf244943b0a16c3f8e3e29ac : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"633c914561e803f6e8a07b91274f04224d5597e0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-06-26 05:48:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6152e1848f01c2ff8ee65725951cd21f888ddf12","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-26 06:27:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"23b9eac9433f643deb857fb2390d1cadb27f6892","revisions":{"a78e270a8de6ba5cd99a8fc9437591daab4a42ea":{"kind":"REWORK","_number":1,"created":"2025-06-26 01:22:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/953370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/70/953370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/70/953370/1"}}},"commit":{"parents":[{"commit":"d91fac698439241be8eb5bb290ad40aa20fe7d8d","subject":"Merge \"Support Nova ephemeral disks in flavor-based reservations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d91fac698439241be8eb5bb290ad40aa20fe7d8d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 01:21:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 01:22:05.000000000","tz":540},"subject":"Update supported python versions","message":"Update supported python versions\n\nPython 3.9 was removed from the tested runtimes in this cycle[1].\nPython 3.8 should have been removed in 2024.2 release.\n\nAlso declare support for Python 3.12 which has been tested for some\ntime.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n\nChange-Id: Ie029a034db0c5448fb3e0388763dfcd3f4b4469c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a78e270a8de6ba5cd99a8fc9437591daab4a42ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a78e270a8de6ba5cd99a8fc9437591daab4a42ea"}]},"branch":"refs/heads/master"},"5bff93e2b6ba32d8520bd607e6f751da38040c01":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-06-26 01:22:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/953370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/70/953370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/70/953370/2"}}},"commit":{"parents":[{"commit":"d91fac698439241be8eb5bb290ad40aa20fe7d8d","subject":"Merge \"Support Nova ephemeral disks in flavor-based reservations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d91fac698439241be8eb5bb290ad40aa20fe7d8d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 01:21:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 01:22:22.000000000","tz":540},"subject":"Update supported python versions","message":"Update supported python versions\n\nPython 3.9 was removed from the tested runtimes in this cycle[1].\nPython 3.8 should have been removed in 2024.2 release.\n\nAlso declare support for Python 3.12 which has been tested for some\ntime and is mandatory now.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n\nChange-Id: Ie029a034db0c5448fb3e0388763dfcd3f4b4469c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5bff93e2b6ba32d8520bd607e6f751da38040c01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5bff93e2b6ba32d8520bd607e6f751da38040c01"}]},"branch":"refs/heads/master"},"23b9eac9433f643deb857fb2390d1cadb27f6892":{"kind":"REWORK","_number":3,"created":"2025-06-26 02:13:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/953370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/70/953370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/70/953370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/70/953370/3"}}},"commit":{"parents":[{"commit":"d91fac698439241be8eb5bb290ad40aa20fe7d8d","subject":"Merge \"Support Nova ephemeral disks in flavor-based reservations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d91fac698439241be8eb5bb290ad40aa20fe7d8d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 01:21:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-26 02:13:36.000000000","tz":0},"subject":"Update supported python versions","message":"Update supported python versions\n\nPython 3.9 was removed from the tested runtimes in this cycle[1].\nPython 3.8 should have been removed in 2024.2 release.\n\nAlso declare support for Python 3.12 which has been tested for some\ntime and is mandatory now.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2025.2.html\n\nChange-Id: Ie029a034db0c5448fb3e0388763dfcd3f4b4469c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/23b9eac9433f643deb857fb2390d1cadb27f6892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/23b9eac9433f643deb857fb2390d1cadb27f6892"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
