)]}'
{"id":"openstack%2Fblazar~953763","triplet_id":"openstack%2Fblazar~master~I2bc3a39313e2534ca0a5638e08fcac736bac66ed","project":"openstack/blazar","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-10 15:34:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2bc3a39313e2534ca0a5638e08fcac736bac66ed","subject":"sqlalchemy: Use built-in declarative","status":"MERGED","created":"2025-06-30 13:20:49.000000000","updated":"2025-07-10 15:35:56.000000000","submitted":"2025-07-10 15:34:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953763","meta_rev_id":"af81e27af3895f6c9ddf552d81de915fc7a65896","_number":953763,"virtual_id_number":953763,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-10 15:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-10 15:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-07-10 15:09:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-07-10 15:09:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-30 13:58:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 15:08:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-07-10 15:09:15.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"108a3ce1f5680e1bc2b9d7c3d7121e9dbbfa3e0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 13:20:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec0409fb2736d784534e56582d00136d5ccd99ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 13:58:00.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b42d38f63c1f483f89995c7e9e3fda81\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8da31673ec34c14a7285d7b4fbc26c6 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4426e02fef43cb91f350221d02215f : FAILURE in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edba0239165d463092c51048174c16db : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c8dc9530c2d4d349982ad371e302db4 : SUCCESS in 6m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb6634a925f6491497a1cd096cbd7503 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbb7cfe0b3444b29e0674f8be284b21 : SUCCESS in 8m 02s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/5623bfff48fd4dc988e6907e88a7c87b : SUCCESS in 35m 41s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/922ad9461b7747e2af8fcd109fe313a0 : SUCCESS in 33m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/41b0d0c22650460f9478a70be096abf9 : FAILURE in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82e8ba404ecaedd3b22a51faa342b612b0a4df01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-30 14:10:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cbbaa3c896f1e5cef11a6715332f8075ce54a2af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 14:57:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f23c75e28964e05a341c57e8572d74f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c979cffb306749ad80dd3640d5c6d00e : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f351e27397d549ce9f2d2b678fcff89c : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b64bf897d5d24fe3b31fd6521ef5c761 : SUCCESS in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b413758dc0ed4e76be6012de8b513f02 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe44e4a4e87d4397a1d770fde3a9e761 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29906e1cd51f4dba9c81ac5bf6505bb8 : SUCCESS in 5m 22s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/29180544cb684ad1a8d14bafc001e14d : SUCCESS in 38m 04s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/9c34b4438c6f4a55813f39f7daf31fc1 : SUCCESS in 41m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b870ec1cba0c48a0a26c901e1f6f52df : FAILURE in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf7a26f73c9bccd91e298b4d19d7598a7db2b3c9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-07-10 15:08:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2f01b88e1d6595afe6d4f3e6478ac14be7116b01","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-07-10 15:09:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a9e2da2386b5372b0b31974c302ced36ac399ca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:09:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6122e974b9bd448d2fa30bba23d7f05b13b60d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:34:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dc9714f33fc4a459e375eab0b42429e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c3132fded94d79a9c6fbf258337644 : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/076dbf5e3b6a49968f1ee066f5095bb2 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d60c9ded06249c3a059bd2d6031f12d : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b35fb7ed43454f81ef20f57ac9130d : SUCCESS in 4m 56s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c592518f0a274b60ad967bbfafaf17f8 : SUCCESS in 20m 57s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/9ca33aa064834f26995cd7b4459ce8dc : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":2},{"id":"87bfc2e4207f5611d51b2b7cd92d6999e999fe5e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:34:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"af81e27af3895f6c9ddf552d81de915fc7a65896","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:35:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b5284599b2f4ac18dda55f954be2696\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e655d801f27c4c98b606895190dd6b71 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c2ced4020244b9fae112a42c9c7f055bc30f2093","revisions":{"a678b608c514beee1920370a4e9478cfe4e20885":{"kind":"REWORK","_number":1,"created":"2025-06-30 13:20:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/953763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/953763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/953763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/953763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/953763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/953763/1"}}},"commit":{"parents":[{"commit":"d91fac698439241be8eb5bb290ad40aa20fe7d8d","subject":"Merge \"Support Nova ephemeral disks in flavor-based reservations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d91fac698439241be8eb5bb290ad40aa20fe7d8d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 13:20:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 13:20:39.000000000","tz":540},"subject":"sqlalchemy: Use built-in declarative","message":"sqlalchemy: Use built-in declarative\n\nsqlalchemy.ext.declarative was deprecated in sqlalchemy 1.4.0, due to\nthe built-in implementations[1].\n\n[1] https://github.com/sqlalchemy/sqlalchemy/commit/450f5c0d6519a439f40\n\nChange-Id: I2bc3a39313e2534ca0a5638e08fcac736bac66ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a678b608c514beee1920370a4e9478cfe4e20885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a678b608c514beee1920370a4e9478cfe4e20885"}]},"branch":"refs/heads/master"},"c2ced4020244b9fae112a42c9c7f055bc30f2093":{"kind":"REWORK","_number":2,"created":"2025-06-30 14:10:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/953763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/63/953763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/953763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/953763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/63/953763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/63/953763/2"}}},"commit":{"parents":[{"commit":"d91fac698439241be8eb5bb290ad40aa20fe7d8d","subject":"Merge \"Support Nova ephemeral disks in flavor-based reservations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d91fac698439241be8eb5bb290ad40aa20fe7d8d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 13:20:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-30 14:09:55.000000000","tz":540},"subject":"sqlalchemy: Use built-in declarative","message":"sqlalchemy: Use built-in declarative\n\nsqlalchemy.ext.declarative was deprecated in sqlalchemy 1.4.0, due to\nthe built-in implementations[1].\n\n[1] https://github.com/sqlalchemy/sqlalchemy/commit/450f5c0d6519a439f40\n\nChange-Id: I2bc3a39313e2534ca0a5638e08fcac736bac66ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c2ced4020244b9fae112a42c9c7f055bc30f2093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c2ced4020244b9fae112a42c9c7f055bc30f2093"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
