)]}'
{"id":"openstack%2Fblazar~958293","triplet_id":"openstack%2Fblazar~master~I5905d97d85955113d07b418cbe65fbafe132cd4b","project":"openstack/blazar","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-05 15:18:39.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-05 16:45:13.000000000","reason":"Change was submitted"},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2026-02-05 16:45:13.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-02-05 15:16:13.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I5905d97d85955113d07b418cbe65fbafe132cd4b","subject":"Add support for traits in flavor plugin","status":"MERGED","created":"2025-08-22 16:22:54.000000000","updated":"2026-02-05 16:46:14.000000000","submitted":"2026-02-05 16:45:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958293","meta_rev_id":"868c19f4d2cdfab159d1f1a1a24891af3655e67e","_number":958293,"virtual_id_number":958293,"owner":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-05 16:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-05 15:18:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-02-05 15:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-05 15:18:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-02-05 15:16:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-02-05 15:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-22 17:12:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-26 08:20:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-08-27 13:40:33.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-08 15:17:24.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"b1cefc311b118c27ffd0f4aab1bdfa665bfb55ee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-08-22 16:22:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7de79e748487c88ebeeed92dbcdbc74be611092b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-08-22 16:26:23.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"ea4a3e44dd2c545974787389976693809dfbacd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 17:12:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f48af50bce7645ec851a39bc971b7b12\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98865dab6b6d4808b9c98363671cff07 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227ac63fa7a14ca1b569998be6935eb3 : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1142e07dea91484f92fe018b8c92760c : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99648d5dd6ef47c5952fb0ce9ccc1d11 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61a2a4167bce41a9b48256c46e216ea1 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93de8da91e3745979d858f13e6ab2257 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f524a8b731114fc396cd73985caa3c8a : SUCCESS in 2m 45s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e673c2cd280a4214a8b10bb1489d2009 : SUCCESS in 42m 14s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/620a9ff9cd5e41e79786469e616c1bc7 : SUCCESS in 43m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5c3b51720ca04df590bb0a9b7aa7120f : FAILURE in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e3af6bf0f750aba96fbbcf7b26d7201e99faa07","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-08-27 13:40:33.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81948531fd1661bf1b05e26a0b4b0d7efc468136","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-08-28 11:10:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7e2ac77026ca8ee89f6df57d911dfbd2ca29f3ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-08-28 21:43:42.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dbd0e52e26bf347eeb6e8f83b3b83a38fbbe0285","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-08-28 21:46:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4f5cd19807dd954da63bce82cd5a5e1b4c512db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-28 22:27:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6904509665e406fa6b70431deebd77b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daeffefe25ee47f893ce3e70228bb9c7 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3008ec41265a477aa93aa9ff592a2a02 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/923acb6749d24a1d9514374635039db2 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f22ebd4fe9814ff4a22892b239540f8b : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/53e43236ce064dcd85d7024c6507c200 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c079c27a4ea41cd97daf8f312d46a6d : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60b93147b7564db09353732879efa490 : SUCCESS in 2m 26s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/4aa0f8997bcb41c28964177042c32b0d : SUCCESS in 41m 12s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/236b22422419497ab061fce7c1a3afc2 : FAILURE in 7m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7c38837d8b5d4824852fca38d12207cf : FAILURE in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46a749c6052bc4c6a1e6c9cbc0af44574c20654d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-09-01 08:29:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"946e1d213a17ce889a163503c66c9d1271d7e543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-01 09:10:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3527f20dcc584f3b8abab9236a31b3bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e03d093d4454a9fa6bd38eb6c610221 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c2deae7e834dff839fc85cade5fe9e : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46f2841dc04d42ddb51122fbf7f88d87 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c66c1d8b447e46fc889171d6ffaa6b37 : SUCCESS in 6m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be0bfbdcd8cc4c959012a65d54c59dfd : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b362be77c6e749269f9e65866af7dc67 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a0816c4ec3946e9a4888323f036d268 : SUCCESS in 3m 26s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/15d2fa83df614226b70dcc63f7852744 : SUCCESS in 26m 47s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/59f1cd35fa74440492342070cbc7a48e : SUCCESS in 39m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ccb24929d3664f4a8055f3f73b26ded4 : FAILURE in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"749e42f9a70229cd78fcd26bddddecc6ce11d6ea","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-09-18 15:11:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"381c12580423d7ed3fc1fe126fa2aa1c3d4ea206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-10-02 16:56:58.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"27104adc00e9e1cac0bbbf53f83b547ffa877c07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-10-02 16:58:08.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"74bb6c311dc7344bb8ca323e4078b7e515b43eee","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-10-02 16:59:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5cf591ea4a0919412af0c44c3620a309c2aecc8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-02 17:33:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bdd3d2ff7264305997fdbb2f41eff32\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a459ad2354094a7dbadd70d18735393b : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6760229a9a83427b93d62cdd70c75ae4 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c7bd0dd94d14c76bf13bc2624c248b2 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62ab81a0f0024c11805f70f93f9c0dc9 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55dd3dd642554b24ae8aa6a0a601ff39 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ed720af40249dc9f07721cda2e4b2e : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8582c720a174baf9899644036c5fb3e : SUCCESS in 3m 07s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/1f2ab735fd2941819d5acffaf88ed622 : SUCCESS in 22m 11s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/64f35f609a9f475a8ec42ba07643abed : SUCCESS in 33m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e4ab52858d704e8f8d5254bc7fe11bcc : FAILURE in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ee6943724fa02e56f651e522049317299f03ed5a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-26 17:05:37.000000000","message":"Patch Set 4: Code-Review-1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f71ae1ac61e409511134b9d6244c9bdda87f8480","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-26 17:38:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b12f7ffe6ea4ee4ad32a57ef18fed18338fa6af8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-27 10:23:53.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7197214d27e2e7673340d0e212aa6065185e6d1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-12-01 21:32:38.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e5040504362ae5615ca776eed1971f78ab0061e5","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2025-12-01 21:35:58.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2e700311fdb0536fa538aca390711c804a3983d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 22:21:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b82cc1446874564862ec52ba115d936\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86110cb21c704829b638c3ec6660b77c : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32502ef2921447de8e8ccc4f59d222d0 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d14fd837bdca4a2e8d621ddb5a804747 : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e50702d0805a452fb7d63af1b67d7955 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03e9e5f4c14a4c54a6de290c1c889c14 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc039db67b764a72bbf48203db5bd95e : SUCCESS in 4m 29s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/267f148e846c4f1cb239b44639716ee5 : SUCCESS in 40m 23s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/56fe4ec986ed478c973371757115db03 : SUCCESS in 41m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/05135d2588b54775b04c2e81beb2f7fe : FAILURE in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b26b713504bb311a374ac7c2c39927dd8d6dd79a","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-12-09 14:27:13.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ee1e4dce33e6eedcc6978b54e6738e4c6fa4fe51","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-08 15:17:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fb59ece4d9c8004436cb040e7bc623ebe6d12941","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-05 12:43:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"504508227d9a6c5fb74addbd351cee67891d7055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-05 15:14:42.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a9922635563535bad114640921203f065aaba54d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-05 15:16:13.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"880d27d6b48cf0f403c37138cbad01f8ec96e74d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-05 15:18:39.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6254fa1c431e1684fc441176d90d4a342e140f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 15:47:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/691df5bb846d489b8b0909261b4e1ac5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6011f6587714563b94a65c800d191b9 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2cfd5b4c0f646a8a55f941cadedd114 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c07f6bc18d4440c930e7ef0db690b74 : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4111db812984476d91f1f32465f00535 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38cced3421594892a512bbf18b7ce4cf : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c947dc86ce94c62aa75486d61af70bd : SUCCESS in 3m 37s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c577ac691f5d49d580038fb7c4693043 : SUCCESS in 31m 08s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/e306b9cea6144dde907ebd51214b5aa5 : SUCCESS in 22m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c1796a74a7fa4fe9b29e30536aac7d7a : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2198acfbe8a54c01fee352ce851c96a18d8329af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 15:48:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"26cd4b4bdb459590aec8f15ed567cd70a014de5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:45:13.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e3d46fbfc7b4a5d8613c2748cfab586\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e19fce587543b5bbd96ecc437b0d1a : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/757e6ac3791e49c0ab6962638149a50d : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27e6d268f5434ea19960fd4abd746c1b : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab5ae565b0a44dcb712b0b3ac85ac53 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d9b001d818d4111912115e4ad30cb25 : SUCCESS in 3m 04s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/e4753e70df4c401b9d7213ce46611fe1 : SUCCESS in 48m 52s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/bbad7f8e8ff54defab00e4a9cfcb2c19 : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":6},{"id":"77a212dc806aecd97243fb9f850576ca793ccffc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:45:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"868c19f4d2cdfab159d1f1a1a24891af3655e67e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:46:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23534e77d1aa49c3b9e5a8b163a3a75c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9271575800674887a2eb9b0684ff6080 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/deca820456b34681a5639b7cadabb47e : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cbb8776ce8ac4ee53c0f945c2b6030473d429751","revisions":{"da044d0f2396b5122d0b5eaf1d81f88ab29478d9":{"kind":"REWORK","_number":1,"created":"2025-08-22 16:22:54.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/93/958293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/93/958293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/93/958293/1"}}},"commit":{"parents":[{"commit":"e407e8da0225e319807e96bf709720b0ee2257fb","subject":"Fix owner policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e407e8da0225e319807e96bf709720b0ee2257fb"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-07-18 17:39:01.000000000","tz":-300},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-08-22 16:22:50.000000000","tz":0},"subject":"Add support for traits in flavor plugin","message":"Add support for traits in flavor plugin\n\nPreviously, traits could not be included with a flavor reservation. This\ncommit implements checking flavor traits when querying available hosts.\n\nThis check is done via the placement API, in parallel where possible. As\na result, `_query_available_hosts` is slower when `resource_traits` is\nnon-empty.\n\nChange-Id: I5905d97d85955113d07b418cbe65fbafe132cd4b\nSigned-off-by: Mark Powers \u003cmarkpowers@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da044d0f2396b5122d0b5eaf1d81f88ab29478d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/da044d0f2396b5122d0b5eaf1d81f88ab29478d9"}]},"branch":"refs/heads/master"},"cab9f3fa47561eacc66f7040fc6bd306f96b1af5":{"kind":"REWORK","_number":2,"created":"2025-08-28 21:43:42.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/93/958293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/93/958293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/93/958293/2"}}},"commit":{"parents":[{"commit":"e407e8da0225e319807e96bf709720b0ee2257fb","subject":"Fix owner policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e407e8da0225e319807e96bf709720b0ee2257fb"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-07-18 17:39:01.000000000","tz":-300},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-08-28 21:42:48.000000000","tz":0},"subject":"Add support for traits in flavor plugin","message":"Add support for traits in flavor plugin\n\nPreviously, traits could not be included with a flavor reservation. This\ncommit implements checking flavor traits when querying available hosts.\n\nThis check is done via the placement API, in parallel where possible. As\na result, `_query_available_hosts` is slower when `resource_traits` is\nnon-empty.\n\nChange-Id: I5905d97d85955113d07b418cbe65fbafe132cd4b\nSigned-off-by: Mark Powers \u003cmarkpowers@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cab9f3fa47561eacc66f7040fc6bd306f96b1af5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cab9f3fa47561eacc66f7040fc6bd306f96b1af5"}]},"branch":"refs/heads/master"},"766f59930cd7638ec4781b42e1c4c04ff91b374a":{"kind":"REWORK","_number":3,"created":"2025-10-02 16:56:58.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/93/958293/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/93/958293/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/93/958293/3"}}},"commit":{"parents":[{"commit":"e407e8da0225e319807e96bf709720b0ee2257fb","subject":"Fix owner policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e407e8da0225e319807e96bf709720b0ee2257fb"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-07-18 17:39:01.000000000","tz":-300},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-10-02 16:54:43.000000000","tz":0},"subject":"Add support for traits in flavor plugin","message":"Add support for traits in flavor plugin\n\nPreviously, traits could not be included with a flavor reservation. This\ncommit implements checking flavor traits when querying available hosts.\n\nThis check is done via the placement API, in parallel where possible. As\na result, `_query_available_hosts` is slower when `resource_traits` is\nnon-empty.\n\nCo-Authored-By: Michael Sherman \u003cshermanm@uchicago.edu\u003e\nChange-Id: I5905d97d85955113d07b418cbe65fbafe132cd4b\nSigned-off-by: Mark Powers \u003cmarkpowers@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/766f59930cd7638ec4781b42e1c4c04ff91b374a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/766f59930cd7638ec4781b42e1c4c04ff91b374a"}]},"branch":"refs/heads/master"},"e5dffcacc5456c8a77489256b7136c7fbb8c2224":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-10-02 16:58:08.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/93/958293/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/93/958293/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/93/958293/4"}}},"commit":{"parents":[{"commit":"e407e8da0225e319807e96bf709720b0ee2257fb","subject":"Fix owner policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e407e8da0225e319807e96bf709720b0ee2257fb"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-07-18 17:39:01.000000000","tz":-300},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-10-02 16:57:04.000000000","tz":0},"subject":"Add support for traits in flavor plugin","message":"Add support for traits in flavor plugin\n\nPreviously, traits could not be included with a flavor reservation. This\ncommit implements checking flavor traits when querying available hosts.\n\nThis check is done via the placement API to get resource providers based\non a trait query only when traits are specified on the flavor.\n\nCo-Authored-By: Michael Sherman \u003cshermanm@uchicago.edu\u003e\nChange-Id: I5905d97d85955113d07b418cbe65fbafe132cd4b\nSigned-off-by: Mark Powers \u003cmarkpowers@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e5dffcacc5456c8a77489256b7136c7fbb8c2224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e5dffcacc5456c8a77489256b7136c7fbb8c2224"}]},"branch":"refs/heads/master"},"47f071fa2d2fed4b2feab7c83c0f5e5bfaea0752":{"kind":"REWORK","_number":5,"created":"2025-12-01 21:32:38.000000000","uploader":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"ref":"refs/changes/93/958293/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/93/958293/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/93/958293/5"}}},"commit":{"parents":[{"commit":"e407e8da0225e319807e96bf709720b0ee2257fb","subject":"Fix owner policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/e407e8da0225e319807e96bf709720b0ee2257fb"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-07-18 17:39:01.000000000","tz":-300},"committer":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-12-01 21:32:30.000000000","tz":0},"subject":"Add support for traits in flavor plugin","message":"Add support for traits in flavor plugin\n\nPreviously, traits could not be included with a flavor reservation. This\ncommit implements checking flavor traits when querying available hosts.\n\nThis check is done via the placement API to get resource providers based\non a trait query only when traits are specified on the flavor.\n\nCo-Authored-By: Michael Sherman \u003cshermanm@uchicago.edu\u003e\nChange-Id: I5905d97d85955113d07b418cbe65fbafe132cd4b\nSigned-off-by: Mark Powers \u003cmarkpowers@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/47f071fa2d2fed4b2feab7c83c0f5e5bfaea0752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/47f071fa2d2fed4b2feab7c83c0f5e5bfaea0752"}]},"branch":"refs/heads/master"},"cbb8776ce8ac4ee53c0f945c2b6030473d429751":{"kind":"REWORK","_number":6,"created":"2026-02-05 15:14:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/958293/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/93/958293/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/93/958293/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/93/958293/6"}}},"commit":{"parents":[{"commit":"23338c3cb4caf98f3f868a232a604907b94546af","subject":"Merge \"Fix updating amount in active instance leases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/23338c3cb4caf98f3f868a232a604907b94546af"}]}],"author":{"name":"Mark Powers","email":"markpowers@uchicago.edu","date":"2025-07-18 17:39:01.000000000","tz":-300},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-05 15:14:27.000000000","tz":60},"subject":"Add support for traits in flavor plugin","message":"Add support for traits in flavor plugin\n\nPreviously, traits could not be included with a flavor reservation. This\ncommit implements checking flavor traits when querying available hosts.\n\nThis check is done via the placement API to get resource providers based\non a trait query only when traits are specified on the flavor.\n\nCo-Authored-By: Michael Sherman \u003cshermanm@uchicago.edu\u003e\nChange-Id: I5905d97d85955113d07b418cbe65fbafe132cd4b\nSigned-off-by: Mark Powers \u003cmarkpowers@uchicago.edu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cbb8776ce8ac4ee53c0f945c2b6030473d429751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/cbb8776ce8ac4ee53c0f945c2b6030473d429751"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
