)]}'
{"id":"openstack%2Fblazar~973450","triplet_id":"openstack%2Fblazar~master~I2f98ffe2cfbfb20c1c26be37bf54705b6666d01f","project":"openstack/blazar","branch":"master","attention_set":{},"removed_from_attention_set":{"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2026-02-05 15:31:54.000000000","reason":"\u003cGERRIT_ACCOUNT_33582\u003e replied on the change","reason_account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-02-05 16:19:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f98ffe2cfbfb20c1c26be37bf54705b6666d01f","subject":"Add flavor-based reservations to API reference","status":"MERGED","created":"2026-01-15 10:18:03.000000000","updated":"2026-02-05 16:20:06.000000000","submitted":"2026-02-05 16:19:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973450","meta_rev_id":"9cab1eb61ab77f2af43cbaafecce20c64d0f51f9","_number":973450,"virtual_id_number":973450,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-05 16:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-23 07:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-01-26 10:03:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-02-05 15:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":1,"date":"2026-02-05 15:31:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-15 11:06:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-23 07:49:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-01-26 10:03:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-26 10:03:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"}],"messages":[{"id":"d042aeb6e5ad2a911a1e2bbe3f38f3efa62370d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-15 10:18:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26ebaad047c5757b4af530b5d5c7eff4a234e63c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 11:06:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ad455e40d6e4054a75ccc3e1a4a1e49\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3950db7cb1f64fe2b6ffe6c917016085 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45f49abf5ffb431c9ba8cae1c58bb953 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44f243994f5481b9e24c3a12f5a39b8 : SUCCESS in 4m 07s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/211e527c7d63432f9dd463b9f6a82c6d : SUCCESS in 43m 25s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/115eede13be84ce78223929af3034add : SUCCESS in 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"df5217729a93743b78d608fefc026a2dc9e54b99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-22 15:12:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"461581561fbc078cf7e6f65e771db8ea8e5fa752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 16:23:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b396c024fe094e6581e1ca94cea7d7db\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1779f5d47b9c4064b5aaec52a1e0d5e0 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3962dbf1bad6417e89709b77f917616e : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e366a8583f64a50927985a7fe4a76d9 : SUCCESS in 8m 44s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b06db112fd8f40859c6c3619ac2f1505 : SUCCESS in 55m 58s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/dbac8c17859a4599a9651dccca777518 : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":2},{"id":"37bf90bb94c68f53fecaa9a685e448864696a289","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-23 07:49:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"edd6aa75c2ccf686ea3ff23badd80c5c8a83120c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 10:03:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2caf02c164a9652ce2a8adaf70c6506d3c8fc9a5","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2026-02-05 15:31:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5abf7fb8e4e9eae3b66800ca46b5d5029a6da761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 15:32:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f431df637bfeb40a7ac8d1bac73ff32cf58e829","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:19:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71c4dff9c6364af09fe02dc8ca64bc72\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a0430921dbdf4397a43c4592725736df : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e12b75ac25469ab2abf0121b1bfe6f : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d906e968460b460da924481b0566bac4 : SUCCESS in 9m 51s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ea990e1cb3f04847a154f19d4723c800 : SUCCESS in 23m 32s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/994ea967fe3146d095c1d9c998157f05 : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a4910412d93a7062978341f9e5d145d97e98297b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:19:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9cab1eb61ab77f2af43cbaafecce20c64d0f51f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:20:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a9e79d79ba142ebaaafd088aed4df51\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6180d464d5df432a88197cb0d4438567 : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6329794172644f8fa1d6a419a180732a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1cab52dcb5c244b390752ac3b4a3510d668bbe3","revisions":{"4865135d93dbde86dd6ca76cfaa4c0f0c4322766":{"kind":"REWORK","_number":1,"created":"2026-01-15 10:18:03.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/50/973450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/50/973450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/973450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/973450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/973450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/50/973450/1"}}},"commit":{"parents":[{"commit":"3e7cf24571e425ab2b806fe3419c9c9c59701f94","subject":"Remove pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3e7cf24571e425ab2b806fe3419c9c9c59701f94"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-15 10:03:14.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-15 10:17:59.000000000","tz":0},"subject":"Add flavor-based reservations to API reference","message":"Add flavor-based reservations to API reference\n\nChange-Id: I2f98ffe2cfbfb20c1c26be37bf54705b6666d01f\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4865135d93dbde86dd6ca76cfaa4c0f0c4322766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4865135d93dbde86dd6ca76cfaa4c0f0c4322766"}]},"branch":"refs/heads/master"},"b1cab52dcb5c244b390752ac3b4a3510d668bbe3":{"kind":"REWORK","_number":2,"created":"2026-01-22 15:12:54.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/50/973450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/50/973450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/973450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/973450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/50/973450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/50/973450/2"}}},"commit":{"parents":[{"commit":"3e7cf24571e425ab2b806fe3419c9c9c59701f94","subject":"Remove pylint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/3e7cf24571e425ab2b806fe3419c9c9c59701f94"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-15 10:03:14.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-22 15:12:47.000000000","tz":0},"subject":"Add flavor-based reservations to API reference","message":"Add flavor-based reservations to API reference\n\nChange-Id: I2f98ffe2cfbfb20c1c26be37bf54705b6666d01f\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b1cab52dcb5c244b390752ac3b4a3510d668bbe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/b1cab52dcb5c244b390752ac3b4a3510d668bbe3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
