)]}'
{"id":"openstack%2Fblazar~974697","triplet_id":"openstack%2Fblazar~stable%2F2025.2~Idd3ee49f173a62392eb5aec7ea664921a36d5068","project":"openstack/blazar","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2026-02-05 15:19:47.000000000","reason":"\u003cGERRIT_ACCOUNT_33582\u003e replied on the change","reason_account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-02-05 16:13:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd3ee49f173a62392eb5aec7ea664921a36d5068","subject":"Fix updating amount in active instance leases","status":"MERGED","created":"2026-01-26 11:05:17.000000000","updated":"2026-02-05 16:14:24.000000000","submitted":"2026-02-05 16:13:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974697","cherry_pick_of_change":973410,"cherry_pick_of_patch_set":1,"meta_rev_id":"aee4b3a5c1186a1ce49aee28f36fefb2e022135e","_number":974697,"virtual_id_number":974697,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-05 16:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-28 09:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-02-05 15:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-02-05 15:19:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 11:49:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 09:19:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-01-28 09:30:36.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"}],"messages":[{"id":"ac629f6221e4b05299e60aaffb8efd2159c9da15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-26 11:05:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2a64d57a0a5ea2020a9f408de5323fa796a4ef8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 11:49:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab4bef047a94c32bd8ef1513d686bc0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/951998e38b854b90b0310520f6859af1 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6fe2926ddba4b8fb0a73ac2cb6e8c88 : SUCCESS in 2m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b42485c48cf347c9ac6f465607e1262d : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6380b8bf71443bf92dddb34c7454f35 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/daa846a44b614ea4bf3eee4c272899aa : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73afa118be4c4600b5ff63e2ebb32f07 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc375413be4448f8a53489f8c5d6a550 : SUCCESS in 2m 30s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/b462f792ff4d4fd5a82667b3ecd28cf7 : SUCCESS in 38m 57s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/8c68e6f7d2a54026afd1120d682874fe : SUCCESS in 34m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5b0a2b69c6f846d1afdf171e645261fb : FAILURE in 2m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5821bd34dcb977a324a8b3a44e7fa2be967e19f1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-28 09:19:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"305b9876a9a1764702ce7967fd7626b4794c0196","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2026-02-05 15:19:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe0ec010ed611fc3db45d1355adf7cd06c104fb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 15:20:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"843c95e3213ace0cbd7c036c378adf4f384afb9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:13:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef8b5d8c11db48208854c45b0ea27921\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795dea2e4646405196d4ab71d218a8cf : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e656584f16d43f49a61d1b71fd8a799 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/015c294d6a3345f3846f721eee32aa4d : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9354318fe9c4a51a74d29487c2df409 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d342ade0dc543e7835a32e3b5c79c20 : SUCCESS in 2m 19s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/46098d46cb7544f2857177eba4b02908 : SUCCESS in 42m 30s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/5568bef963e94d8d85daafb47aee831f : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"6fe55caa26de778c782749b3376068ecf66da591","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:13:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aee4b3a5c1186a1ce49aee28f36fefb2e022135e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 16:14:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b3b223ea0874895b3b64e6ad687e692\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c051b274a33421890e255ce2bffe798 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a260283370d44708fe00846aeed082b : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5637b6e3535dab854e7fd69acde13e3433934a04","revisions":{"5637b6e3535dab854e7fd69acde13e3433934a04":{"kind":"REWORK","_number":1,"created":"2026-01-26 11:05:17.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/97/974697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/97/974697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/974697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/974697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/97/974697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/97/974697/1"}}},"commit":{"parents":[{"commit":"49e42f4157fec69268eb1c99da946f2566508ad1","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/49e42f4157fec69268eb1c99da946f2566508ad1"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-14 16:49:16.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-01-26 11:05:17.000000000","tz":0},"subject":"Fix updating amount in active instance leases","message":"Fix updating amount in active instance leases\n\nSince \u0027resource_properties\u0027 was added, \u0027amount\u0027 is no longer the\nlast item in the updatable list, so updating the amount would be\ndenied. Correct this by instead explicitly checking for \u0027amount\u0027.\n\nCloses-Bug: #2138386\nChange-Id: Idd3ee49f173a62392eb5aec7ea664921a36d5068\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n(cherry picked from commit 48d3239171f615d4c1fe8a222e3b9770268a9329)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5637b6e3535dab854e7fd69acde13e3433934a04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/5637b6e3535dab854e7fd69acde13e3433934a04"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
