)]}'
{"id":"openstack%2Fblazar~979873","triplet_id":"openstack%2Fblazar~master~I0d74e02a63cfa479c5a75457edb046e0b9824dd5","project":"openstack/blazar","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-08-28 08:47:56.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"33582":{"account":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"last_update":"2026-08-28 09:47:39.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-28 09:47:39.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-08-28 09:47:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d74e02a63cfa479c5a75457edb046e0b9824dd5","subject":"Replace remaining eventlet usage with AsyncIO","status":"MERGED","created":"2026-03-10 15:51:03.000000000","updated":"2026-08-28 09:48:51.000000000","submitted":"2026-08-28 09:47:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979873-eventlet-removal","meta_rev_id":"91568e234fa9c67eaf29fda6b39ca8cba5d83430","_number":979873,"virtual_id_number":979873,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-28 09:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-28 08:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-28 08:47:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-10 17:35:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-19 15:31:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-04-01 14:52:58.000000000","updated_by":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"reviewer":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"state":"REVIEWER"},{"updated":"2026-06-29 11:39:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-06-29 11:39:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"d5a6c3098555f4463ff476fc3e6073832cccb786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-10 15:51:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d37331578a203a0c4053d41579ae7de7f5ca4c6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-10 16:13:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a290111bef191ab149745a6d64f0e56f2e8ff20e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-10 16:30:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"02866ae2780ef4216e59c7ae540ce0f3a85f9083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 17:35:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c5498e63af4438298f7052e79b41195\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4cd7ef9145d4fc68bfb11b961e74c14 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56182db0f69d4b0a8f8810d28f405bea : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c35ae75d69a548b58a6e968dfa2a7596 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cd16d8707014bee80406ad1d2e7c977 : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5dbabcb9abc745a1bd728b1cf53020e4 : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2e6783b1b04e1788d7b14c1879af09 : SUCCESS in 8m 11s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/8d87381353aa41af9bdfedf0f9d7aa7f : FAILURE in 55m 58s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/cf501a33115147bda7c7252f660678ec : FAILURE in 35m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ab63252d0504e69936b618547c6eb85 : FAILURE in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2eba5fed0d008f47de38adddcfb81af9d63fc08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-12 16:16:47.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9d0fc66f95f49ec179ea61fbdeb03ceebdd574b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 17:12:10.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/05f79d29c9fc472d99b5d9292b7847f3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e805fc8ab95492380ed161b4ccd41ed : SUCCESS in 4m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebca8886c3714b578ea562c98841f355 : FAILURE in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20935bf4449435dac10d8ce1abe30c8 : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d3dcd8f022944699d1556e07ba6b352 : FAILURE in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a5f84642b9c495194972eaf3e06fa8b : FAILURE in 6m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/20c2899e888e4970a3d9a6ce070ad49f : FAILURE in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3adacbb8bee426b91b2f6b5d9d9858d : SUCCESS in 8m 55s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/334524c667484050b3e3ee38e550a140 : FAILURE in 38m 48s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/b63ea067e2b5431fba546b0de1ebbd7a : FAILURE in 37m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0e8ad39f0d9f4bf19fbe7935445409c2 : FAILURE in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bce03d5ed9da5b6a836e26f009f245c15a5b7ccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-17 15:46:07.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f2674674fdee6387efe3f97943b94997a1e26fab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-17 15:47:46.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"90d4d31e5343bffb19471732d076274282604680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-17 15:50:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"48b9d7c173db974493411e3acdaec8ded6804098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 16:39:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9e48693ef449f6914a11d06ffc8d7e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/08f7ddaaab8a4bce99821375ae9cef23 : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce6d0160c9074bf183f242d0fe838945 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddf6742ff20a45d6ad9573f2563d6631 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86b397d5cb3a47739ac786006f41538e : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0252a58893a84ff1a34ea8f0232b6d4d : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a2f7929f1154eb9a5cf82c9e91be04f : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c21cbd35834e4995a46eb28335eafb84 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9dc69170b84035a1cc1a8e6bdddbcc : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/422d3ece539b4a0e8b6ad7513ab95b1e : SUCCESS in 3m 07s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c59ee349eab7421fab39f7b286779e44 : SUCCESS in 40m 53s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/c1270e1280404f0b9aae1f127817a91b : SUCCESS in 24m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42b10c00d0dc44608bacfe72f7bafa67 : FAILURE in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b4809323fd0653ac3befcce045ee2cc638fe6f25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-17 16:40:47.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"107419e882231f54195fa8c7fb414af072bab1e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 17:30:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6e0ef8adc894440bda08df3dc54852f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7f26d6ce8c5e4b0f87d74db7f6999dd7 : SUCCESS in 2m 44s\n- requirements-check https://zuul.opendev.org/t/openstack/build/33784df5c87e48b0aa486000b488308f : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68c0325c11ab430d8311be57d91afd46 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58d2d3f0f6434a76ad4d2c69de1a8cd6 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97ac0faa92314c3ca7d2426fd0a801c7 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83354115fd804e01944b956cb91c9e79 : SUCCESS in 5m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f250133921314b5fbfe03e93cbfa6c07 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55817a22d79a40b986bae4e8e7582393 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e58625d319b643e1966a10ee4083fa54 : SUCCESS in 2m 33s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/26691626aacf4220ac28ebae75a987dc : SUCCESS in 46m 03s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/3046009e6d5247389c988cf124d9863a : SUCCESS in 33m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d36b2b9d0ae241efad2aa320ecc8536b : FAILURE in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"eb00f631b9e69e3ae202a4dba4e927aae17a8aed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-19 15:31:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d5f1519061c1472ceb99429c68afc5323df04bf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-30 12:00:05.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"c56e29330953805ba9ef8abe2a4a2d2bda6ef8d3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-30 12:00:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1850e6d2488fce5d0920acab8b455e2fb18e61bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 12:35:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebee5ae02ee34a479f781ebbc01548de\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f9c91151c6b4a35b246a2c5bd370447 : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a91089d8fc0449afae275e5685ae71e5 : SUCCESS in 1m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4b3e37f547a44e1838fe5ca1a7decfc : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e792ceba19147f5aa56738c2bdf32f5 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/873dcfec61434775b16f954d5c5b93dc : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a093cfad2e34b12bca2b0f7c173e07d : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1278e713d837466380c886c20f8cb4ce : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee767bf52eae41d0b78168afde6b06af : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81eca07cd539403495e5aae5770b32f0 : SUCCESS in 3m 19s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/c2b776af502d4637b7866d60b4ad3b4a : SUCCESS in 25m 50s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f299c508f84a48f38769838114e9a15f : SUCCESS in 33m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4068b6407f6a409b96b90c0da936de6f : FAILURE in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"415271e041042bb237246e2e3726c7ee03eda53f","author":{"_account_id":33582,"name":"Mark Powers","email":"markpowers@uchicago.edu","username":"markpowers"},"date":"2026-04-01 14:52:58.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a9818c81f08b9cd459a9594c41d10796419d0103","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-07 09:52:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"02a6eef2f390c419bf4769d35e8b00b1f0bc9621","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-29 11:39:11.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"e5d37742705735b258b66a7888e38201d7f9b4e3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-25 13:23:59.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9b464e98d56c2e6cc12af46e16fb57bdf35e8235","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-08-25 15:00:46.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"952155f58a77bee8788eea91559bd55269106524","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-08-25 15:01:02.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"179c6fd28bed3c56fbe647275c2e26ce5460b793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 15:51:26.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ca101def6834aeaac30bfabe11b4e5f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ba2cbceab084480b3d32131ff45933c : FAILURE in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a20548221109462dbec481a55cc4f024 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a95d1d405444a2927ad50e9c7200b5 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90fa16d3acaa44cdab0e8e5777fcfa03 : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95c28bb31c7a4cfd95d3c8089cc2a4b2 : SUCCESS in 6m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/903b1370285b40f4b847414c66dee7a1 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e48a9d54d048cb83c69527ff6ba96c : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c65ee4239b994793936e0e0b778365a9 : SUCCESS in 5m 35s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/d0ffc696427b4579afdc5623fcc3ba9f : SUCCESS in 43m 20s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/1b2e44cac5f54bdd8a7f23fc04869a19 : SUCCESS in 41m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/730d0ccceec54999bc9f9545fb1262cd : FAILURE in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ebc0f72695f0226468d4d67fafc0a8de5443ea5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-28 07:53:31.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_35263\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"_revision_number":11},{"id":"b853f3f67b349839a40bf3cf3c000f740f9a1012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 08:36:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/357d3d2d36904b3f969126f5f08f2001\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d4fc4d479da74390abf751479df19128 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6712ea39c987486497ef27c734e4ebcf : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e579cc1f1af409c8984504ca0827ba9 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/138aeb9cb9ee41b49da0db4380daeb08 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eec5b274012643439fd420bfe3f5ac43 : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fcab19042943461fac152ababdd0fcc1 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eed0a0926c44745bfbf6def181de1fc : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5f75cd1f65b499aa65ce33c3853666b : SUCCESS in 4m 21s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/57457c7163ec4c56aa6bc96b5eaad8a3 : SUCCESS in 22m 59s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/f4b6df365f2549768c90a29d66186803 : SUCCESS in 41m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ac8c06088271405881fb5f8578e8a2d7 : FAILURE in 3m 46s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"42872055f91d6be50d4b63b67b775c00af8a9ad4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-08-28 08:47:56.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5fba70cbc70b7ecf9c51eab032940d86a73ff1d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 08:48:34.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"83efb49603c6422933447be00d26496db628b41c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 09:47:38.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7647e48900be42768d8a876604d4bc92\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/571aa77d787e43338530a17aeda73bec : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9cf0cde6ff4825ae6e7769446b8f03 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/263b1fc90b594b42a4699ab02335f45a : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/223a54cb97c24881a9300a2484c22f9a : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dca54509c7fe4eb2bb5a5ed85f076380 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f306c260aaa467d9fa18b4d23bb9794 : SUCCESS in 4m 53s\n- blazar-tempest-plugin-base https://zuul.opendev.org/t/openstack/build/ec782012fcc0498a9c8b5ff1994cc5c4 : SUCCESS in 42m 47s\n- blazar-tempest-plugin-ipv6-only https://zuul.opendev.org/t/openstack/build/a70c1636ff2a4502bd7e1786ec2e4b9d : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":11},{"id":"e8e05a6bbcefa9400c451c0d6c4d551fc0a9b1df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 09:47:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"91568e234fa9c67eaf29fda6b39ca8cba5d83430","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 09:48:51.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d21c5e8c2524dea98f095e99183515c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1191769d6f4734b083a074dfa2c1ad : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"bbce338f28e4603f9caec8431be239b6cb17f936","revisions":{"83bb360d7d0cc5f2d6c21ce972b043782f3ff9e4":{"kind":"REWORK","_number":1,"created":"2026-03-10 15:51:03.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/1"}}},"commit":{"parents":[{"commit":"ebd80979cec1ccf830c4b4b10bb28cffd62d2902","subject":"WIP: Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ebd80979cec1ccf830c4b4b10bb28cffd62d2902"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:50:55.000000000","tz":0},"subject":"WIP: Replace eventlet with AsyncIO in manager","message":"WIP: Replace eventlet with AsyncIO in manager\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/83bb360d7d0cc5f2d6c21ce972b043782f3ff9e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/83bb360d7d0cc5f2d6c21ce972b043782f3ff9e4"}]},"branch":"refs/heads/master"},"49eb71fd3aa3da2e8fdd4561a34d7937095f5afc":{"kind":"REWORK","_number":2,"created":"2026-03-10 16:13:36.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/2"}}},"commit":{"parents":[{"commit":"ebd80979cec1ccf830c4b4b10bb28cffd62d2902","subject":"WIP: Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ebd80979cec1ccf830c4b4b10bb28cffd62d2902"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 16:13:28.000000000","tz":0},"subject":"WIP: Replace eventlet with AsyncIO in manager","message":"WIP: Replace eventlet with AsyncIO in manager\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/49eb71fd3aa3da2e8fdd4561a34d7937095f5afc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/49eb71fd3aa3da2e8fdd4561a34d7937095f5afc"}]},"branch":"refs/heads/master"},"030d262d2814fcc219e01d3b71543bb9dcc248a7":{"kind":"REWORK","_number":3,"created":"2026-03-10 16:30:42.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/3"}}},"commit":{"parents":[{"commit":"ebd80979cec1ccf830c4b4b10bb28cffd62d2902","subject":"WIP: Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ebd80979cec1ccf830c4b4b10bb28cffd62d2902"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 16:30:34.000000000","tz":0},"subject":"WIP: Replace eventlet with AsyncIO in manager","message":"WIP: Replace eventlet with AsyncIO in manager\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/030d262d2814fcc219e01d3b71543bb9dcc248a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/030d262d2814fcc219e01d3b71543bb9dcc248a7"}]},"branch":"refs/heads/master"},"a960b77b16be271d04e904c683048b5a9a9463c4":{"kind":"REWORK","_number":4,"created":"2026-03-12 16:16:47.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/4"}}},"commit":{"parents":[{"commit":"ebd80979cec1ccf830c4b4b10bb28cffd62d2902","subject":"WIP: Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/ebd80979cec1ccf830c4b4b10bb28cffd62d2902"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-12 16:16:04.000000000","tz":0},"subject":"WIP: Replace eventlet with AsyncIO in manager","message":"WIP: Replace eventlet with AsyncIO in manager\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a960b77b16be271d04e904c683048b5a9a9463c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a960b77b16be271d04e904c683048b5a9a9463c4"}]},"branch":"refs/heads/master"},"d617779d2c961a3e65e3cef74a103e84679d3aa9":{"kind":"REWORK","_number":5,"created":"2026-03-17 15:46:07.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/5"}}},"commit":{"parents":[{"commit":"d5fc888c7e14c5c33fda08c755c2f03d66132e4a","subject":"Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d5fc888c7e14c5c33fda08c755c2f03d66132e4a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-17 15:46:00.000000000","tz":0},"subject":"Replace remaining eventlet usage with AsyncIO","message":"Replace remaining eventlet usage with AsyncIO\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d617779d2c961a3e65e3cef74a103e84679d3aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d617779d2c961a3e65e3cef74a103e84679d3aa9"}]},"branch":"refs/heads/master"},"a8fe570b1b70eae03eeb939b3b8d33aeb6a86433":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-03-17 15:47:46.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/6"}}},"commit":{"parents":[{"commit":"d5fc888c7e14c5c33fda08c755c2f03d66132e4a","subject":"Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d5fc888c7e14c5c33fda08c755c2f03d66132e4a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-17 15:47:09.000000000","tz":0},"subject":"Replace remaining eventlet usage with AsyncIO","message":"Replace remaining eventlet usage with AsyncIO\n\nThis also moves to use Threading backend for oslo.service,\nreplacing the deprecated Eventlet backend.\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a8fe570b1b70eae03eeb939b3b8d33aeb6a86433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/a8fe570b1b70eae03eeb939b3b8d33aeb6a86433"}]},"branch":"refs/heads/master"},"c6a7ec3d2d40045499285c41f00ba360d81e1365":{"kind":"REWORK","_number":7,"created":"2026-03-17 15:50:12.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/7"}}},"commit":{"parents":[{"commit":"d5fc888c7e14c5c33fda08c755c2f03d66132e4a","subject":"Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d5fc888c7e14c5c33fda08c755c2f03d66132e4a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-17 15:50:01.000000000","tz":0},"subject":"Replace remaining eventlet usage with AsyncIO","message":"Replace remaining eventlet usage with AsyncIO\n\nThis also moves to use Threading backend for oslo.service,\nreplacing the deprecated Eventlet backend.\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c6a7ec3d2d40045499285c41f00ba360d81e1365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/c6a7ec3d2d40045499285c41f00ba360d81e1365"}]},"branch":"refs/heads/master"},"82e91c20643696ad4428aed0e74edbb9a49b09b4":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-03-17 16:40:47.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/8"}}},"commit":{"parents":[{"commit":"d5fc888c7e14c5c33fda08c755c2f03d66132e4a","subject":"Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d5fc888c7e14c5c33fda08c755c2f03d66132e4a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-17 16:40:47.000000000","tz":0},"subject":"Replace remaining eventlet usage with AsyncIO","message":"Replace remaining eventlet usage with AsyncIO\n\nThis also moves to use Threading backend for oslo.service and\noslo.messaging replacing the deprecated Eventlet backend.\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/82e91c20643696ad4428aed0e74edbb9a49b09b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/82e91c20643696ad4428aed0e74edbb9a49b09b4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"003fb7aaee87efdc87cf34da362f37603ee2beb5":{"kind":"REWORK","_number":9,"created":"2026-03-30 12:00:05.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/9"}}},"commit":{"parents":[{"commit":"d5fc888c7e14c5c33fda08c755c2f03d66132e4a","subject":"Replace eventlet with threading in API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/d5fc888c7e14c5c33fda08c755c2f03d66132e4a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-30 11:59:58.000000000","tz":60},"subject":"Replace remaining eventlet usage with AsyncIO","message":"Replace remaining eventlet usage with AsyncIO\n\nThis also moves to use Threading backend for oslo.service and\noslo.messaging replacing the deprecated Eventlet backend.\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/003fb7aaee87efdc87cf34da362f37603ee2beb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/003fb7aaee87efdc87cf34da362f37603ee2beb5"}]},"branch":"refs/heads/master"},"bd96d7842f9a3ea0afc247db1a06eb49c56b9868":{"kind":"REWORK","_number":10,"created":"2026-08-25 15:00:46.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/73/979873/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/10"}}},"commit":{"parents":[{"commit":"4afcd410015108d7740bece8ee2595e11d94e6e7","subject":"Merge \"Replace eventlet with threading in API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/4afcd410015108d7740bece8ee2595e11d94e6e7"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-08-25 15:00:39.000000000","tz":60},"subject":"Replace remaining eventlet usage with AsyncIO","message":"Replace remaining eventlet usage with AsyncIO\n\nThis also moves to use Threading backend for oslo.service and\noslo.messaging replacing the deprecated Eventlet backend.\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bd96d7842f9a3ea0afc247db1a06eb49c56b9868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bd96d7842f9a3ea0afc247db1a06eb49c56b9868"}]},"branch":"refs/heads/master"},"bbce338f28e4603f9caec8431be239b6cb17f936":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-08-28 07:53:31.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"real_uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/979873/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/blazar","ref":"refs/changes/73/979873/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/blazar refs/changes/73/979873/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/blazar refs/changes/73/979873/11"}}},"commit":{"parents":[{"commit":"0c33c301e2a24fe28a4ea99597b206b87d748022","subject":"Migrate setup configuration to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/0c33c301e2a24fe28a4ea99597b206b87d748022"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-03-10 15:49:42.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-08-28 07:53:31.000000000","tz":0},"subject":"Replace remaining eventlet usage with AsyncIO","message":"Replace remaining eventlet usage with AsyncIO\n\nThis also moves to use Threading backend for oslo.service and\noslo.messaging replacing the deprecated Eventlet backend.\n\nChange-Id: I0d74e02a63cfa479c5a75457edb046e0b9824dd5\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bbce338f28e4603f9caec8431be239b6cb17f936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/blazar/commit/bbce338f28e4603f9caec8431be239b6cb17f936"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"4afcd410015108d7740bece8ee2595e11d94e6e7","ours":"bd96d7842f9a3ea0afc247db1a06eb49c56b9868","theirs":"0c33c301e2a24fe28a4ea99597b206b87d748022","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
