)]}'
{"id":"openstack%2Fceilometer-powervm~678504","triplet_id":"openstack%2Fceilometer-powervm~master~Ic60993c2315e8b10da8abb1fb02e33151a1e2b98","project":"openstack/ceilometer-powervm","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ic60993c2315e8b10da8abb1fb02e33151a1e2b98","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-08-26 09:37:35.000000000","updated":"2019-09-13 04:35:53.000000000","submitted":"2019-09-13 04:35:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678504-1568349353369-b9dbb748","meta_rev_id":"3a04552ee0f290564544ffcbb1602936e7b6c656","_number":678504,"virtual_id_number":678504,"owner":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-09-13 04:35:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-11 13:00:26.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"all":[{"value":2,"date":"2019-09-09 10:23:50.000000000","_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":2,"date":"2019-08-26 16:44:45.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"all":[{"value":1,"date":"2019-09-09 10:24:03.000000000","_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-26 16:44:45.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-09 10:24:03.000000000","updated_by":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"reviewer":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"state":"REVIEWER"},{"updated":"2019-09-11 13:00:26.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2019-09-13 04:35:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb6728017332d75292ba0d4e6a03bb1339461549","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-08-26 09:37:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab08974b727dac6c618734e22f1ce9814f984fa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 09:57:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35aa83d8f2294b5a856bb00defa0cfff : SUCCESS in 14m 43s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d7038d352ca1414cba7cc6abbd69f22e : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2505c28653f4bc1a34035d61f72df48 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c4b8fe0797a413f9f144fcf2db93d2b : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7b7917a255a4e71b8ad6f24fdf6c046 : POST_FAILURE in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/583613842c1a4b038e6c6bf837fe6d57 : SUCCESS in 13m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7278123685094ae593c854492914b14a : FAILURE in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"31291b64a1baca4c6d1a2f9a6a1d3b5a1d33ede6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-26 16:01:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a4b6bf1731126baaa446aa1ef244134640cb4abd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:22:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241a29efc9234821a32331d75019cff4 : SUCCESS in 5m 32s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/219821d547964d8f8d9483ab2a820906 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db22dd79c424a02bd3620c533aad191 : SUCCESS in 6m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c1923e9ec162442da8cfca1e889f6334 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adc631b725fb430b874b5f49349110cc : SUCCESS in 6m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf0cb02a651741179e25297ee85344cf : SUCCESS in 6m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9f141cbe2904899871da60ab33123c0 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d156c50f14fa28e891743338ec1adf949c5755a5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-26 16:44:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"76c8f8b6fdef6b3b870e75a747409a5bf0c644fb","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2019-09-09 10:23:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3149ba473baef1799572ee9e3c829376cf860e00","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2019-09-09 10:24:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f6df4686c96371514c9b13e2ef359bfe52e80575","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-11 12:58:07.000000000","message":"Topic set to python3-train","accounts_in_message":[],"_revision_number":2},{"id":"d336eec6f6ed353122b2e971305747bd267c07d2","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-09-11 13:00:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"904468961c55b45345bafc6cae1381be52257dcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 04:23:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6131e5f7cd49b2d18b466599a7c920587f103fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 04:35:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3a04552ee0f290564544ffcbb1602936e7b6c656","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 04:35:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aaee3c129ea4bd8aa07f40098a6d4de : SUCCESS in 5m 44s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a1830e5d2fd42d7a4e5e83b13a53ca9 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/242205e589264af08b65c2899e20569a : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7b0a3e0377944f693dee1310bbef62e : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3d1b8a645364ff2a2648b9552141f61 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/def4a274f2d94f138499f4295ca1ae82 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fdbb384fbb0404992832e7a5ba54ed5 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f3008d1ee22329e0cd8e76aaf8b7b4a2a56af876","revisions":{"1d2ce3c5fd2de2eff95ac2f2f163e7757b46d51d":{"kind":"REWORK","_number":1,"created":"2019-08-26 09:37:35.000000000","uploader":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"ref":"refs/changes/04/678504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/04/678504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678504/1"}}},"commit":{"parents":[{"commit":"ec05201839da26fb7ea746cd15e23088c7e34a74","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/ec05201839da26fb7ea746cd15e23088c7e34a74"}]}],"author":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-08-26 09:36:25.000000000","tz":-240},"committer":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-08-26 09:36:25.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ic60993c2315e8b10da8abb1fb02e33151a1e2b98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/1d2ce3c5fd2de2eff95ac2f2f163e7757b46d51d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/1d2ce3c5fd2de2eff95ac2f2f163e7757b46d51d"}]},"branch":"refs/heads/master"},"f3008d1ee22329e0cd8e76aaf8b7b4a2a56af876":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-26 16:01:37.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/04/678504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/04/678504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678504/2"}}},"commit":{"parents":[{"commit":"edb9284fb2ca4bf77097c92f95c6e721a0aa759a","subject":"Add bindep.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/edb9284fb2ca4bf77097c92f95c6e721a0aa759a"}]}],"author":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-08-26 09:36:25.000000000","tz":-240},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-26 16:01:37.000000000","tz":0},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ic60993c2315e8b10da8abb1fb02e33151a1e2b98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/f3008d1ee22329e0cd8e76aaf8b7b4a2a56af876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/f3008d1ee22329e0cd8e76aaf8b7b4a2a56af876"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
