)]}'
{"id":"openstack%2Fceilometer-powervm~678604","triplet_id":"openstack%2Fceilometer-powervm~master~If87e74917af3bb86b7c166ccc411acce1d6cf6ce","project":"openstack/ceilometer-powervm","branch":"master","topic":"bindep","hashtags":[],"change_id":"If87e74917af3bb86b7c166ccc411acce1d6cf6ce","subject":"Add bindep.txt","status":"MERGED","created":"2019-08-26 16:01:10.000000000","updated":"2019-09-13 04:35:52.000000000","submitted":"2019-09-13 04:35:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678604-1568349352337-900a881e","meta_rev_id":"d56665c8db79f5df3e8488b8c479fcf9961c7308","_number":678604,"virtual_id_number":678604,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":16069,"name":"Gautam Prasad","email":"gautpras@in.ibm.com","username":"gautamd"},{"value":2,"date":"2019-09-13 04:35:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"all":[{"value":2,"date":"2019-09-06 05:54:02.000000000","_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":2,"date":"2019-09-06 06:52:08.000000000","_account_id":16069,"name":"Gautam Prasad","email":"gautpras@in.ibm.com","username":"gautamd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"all":[{"value":1,"date":"2019-09-13 04:23:30.000000000","_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":16069,"name":"Gautam Prasad","email":"gautpras@in.ibm.com","username":"gautamd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"_account_id":16069,"name":"Gautam Prasad","email":"gautpras@in.ibm.com","username":"gautamd"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 06:52:08.000000000","updated_by":{"_account_id":16069,"name":"Gautam Prasad","email":"gautpras@in.ibm.com","username":"gautamd"},"reviewer":{"_account_id":16069,"name":"Gautam Prasad","email":"gautpras@in.ibm.com","username":"gautamd"},"state":"REVIEWER"},{"updated":"2019-09-13 04:23:30.000000000","updated_by":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"reviewer":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"state":"REVIEWER"},{"updated":"2019-09-13 04:35:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1a10ba377d0b28e25285285e63ce61d613e3b87","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-26 16:01:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8821c69df3dce7d14fd75a534c3b60c585d0de2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:17:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29bd427dc79e4361957f63cba662131b : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c7fd9fdf314c4496b22835eb39daad : SUCCESS in 8m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23019bef91ee4a9296b21fc527ebe52d : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c0f72e5fd22413b9eb6649e29ae8a72 : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd2c46a4098b41df84c0e3ab88459490 : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3d7e4483732a0a694b32f95ebe46b17e295cf605","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2019-09-06 05:54:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d09bd9030fd3202a2683e9e8e8ad37b31beb0534","author":{"_account_id":16069,"name":"Gautam Prasad","email":"gautpras@in.ibm.com","username":"gautamd"},"date":"2019-09-06 06:52:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be223e83b3ef2169bd260521ed2b5206de9db260","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2019-09-13 04:23:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"474bf6ffa0319cf1ebea88cb36f67c5ad90dac35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 04:23:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ca5f9586e1d48b0aca776c27131747e0bbc502b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 04:35:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2be88af43b4daa810ff8081dbcef47 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8cc768dd51d4b02a40d35ec2a7d61be : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9d71c5b0b024285a69464b6776174b6 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7bb105f23124796aa976d0abfef987e : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc4b4aa9f3a14e03945d81757a9c9de2 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"94d91fe04acf5ee729e3e9c06e51ab4c93497deb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 04:35:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edb9284fb2ca4bf77097c92f95c6e721a0aa759a","revisions":{"edb9284fb2ca4bf77097c92f95c6e721a0aa759a":{"kind":"REWORK","_number":1,"created":"2019-08-26 16:01:10.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/04/678604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/04/678604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/04/678604/1"}}},"commit":{"parents":[{"commit":"ec05201839da26fb7ea746cd15e23088c7e34a74","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/ec05201839da26fb7ea746cd15e23088c7e34a74"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-26 16:00:44.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-08-26 16:01:09.000000000","tz":-300},"subject":"Add bindep.txt","message":"Add bindep.txt\n\n...with some of the things necessary to get lxml building again.\n\nChange-Id: If87e74917af3bb86b7c166ccc411acce1d6cf6ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/edb9284fb2ca4bf77097c92f95c6e721a0aa759a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/edb9284fb2ca4bf77097c92f95c6e721a0aa759a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
