)]}'
{"id":"openstack%2Fceilometer-powervm~681755","triplet_id":"openstack%2Fceilometer-powervm~master~Ia74d470a6d593a6a81779788e2848c5b5995c44f","project":"openstack/ceilometer-powervm","branch":"master","hashtags":[],"change_id":"Ia74d470a6d593a6a81779788e2848c5b5995c44f","subject":"Enable support for both python3.6 and python3.7","status":"MERGED","created":"2019-09-12 14:02:37.000000000","updated":"2019-10-17 09:35:07.000000000","submitted":"2019-10-17 09:35:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681755-1571304907951-14372de2","meta_rev_id":"d9e3035b8fc49f692efe93a2a57681260f2d5d8c","_number":681755,"virtual_id_number":681755,"owner":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 09:35:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"recommended":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"all":[{"value":1,"date":"2019-10-14 13:43:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-17 09:24:57.000000000","_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"all":[{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-17 09:24:57.000000000","_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-14 15:19:08.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-10-14 13:43:34.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-10-17 09:35:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4bc73a94a9a897fbebc8f2b08633632f2277df0c","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-09-12 14:02:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"561cfee7c96685488e81e59d4dfb984796d0194a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 14:28:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01afc8d89924c0d85f5cb17612df92c : SUCCESS in 3m 26s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2ea7d58c07ef44489f01b9997a758c24 : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc174e10c2f742b791f75877143d2c92 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/517d5f9d84bb4dddb4cf5de1f1e9a729 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2a2599ccf3c499caac0064f0e1f27f6 : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d009c3e170bd424a863ae545b6c376f9 : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"51d4a3fd8c52e5e6da841ca954557349a21f2e62","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-09-13 05:27:54.000000000","message":"Patch Set 1:\n\nreverify","accounts_in_message":[],"_revision_number":1},{"id":"4ee54966ce10b7850d41ab68b2e4b68699189d2d","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-09-14 15:19:08.000000000","message":"Patch Set 1: Code-Review-1\n\nplease fix the conflict first, and the others is ok with me","accounts_in_message":[],"_revision_number":1},{"id":"965d0a5a1323d22e35c1c5aba8fc6c470f54c1c5","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-09-17 03:52:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"589f5023f4ea74bdcbfbec7a7583b7b566185e7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 04:08:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b21bccaf8dae4c4e88777dd8574be472 : SUCCESS in 4m 56s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/281b2b69ddb849d8a5c73dab04139e53 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e41558bb9c2c47d68e872a42155ef059 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5f28b0172b849249d4380ae7f228f53 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79cce51762264571a27c30a8f91e28d1 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34f96dadb42f4296a78d17b63b2fff56 : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c854e73e984419e9fd55f181d62bb61 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5470a2b92d904b533eda2c3e745a26f70730801d","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-10-14 13:43:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a4cc37ddcb0c73def5d671197a1fbe4a176dd863","author":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"date":"2019-10-17 09:24:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"017b98479253e9d0ba9c4657de89c88270acf63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 09:25:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"496a884fed800bc5482a242456de861a637520a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 09:35:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4109e35c567406d8085caeeedb5c2b8 : SUCCESS in 4m 01s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f73cf35e59664dedb23a02ffae473fa7 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052fb23d60f8494ca8757260806f0154 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b511831fd4f64a14a19fddb7c9735495 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14ea57a61ed64df187213b79fe2d6895 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbf091635d284b3791906d2cba1087ac : SUCCESS in 6m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/815464ba81a541458eebfd9cf89e2945 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a8dd0c99387f2d454be7d81ea798c6385f30d094","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 09:35:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7ca211e7007b4444b543457c8627c29ab37483cb","revisions":{"d27e32107047e18f388046b1ff627e894a9b0304":{"kind":"REWORK","_number":1,"created":"2019-09-12 14:02:37.000000000","uploader":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"ref":"refs/changes/55/681755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/55/681755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/55/681755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/55/681755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/55/681755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/55/681755/1"}}},"commit":{"parents":[{"commit":"ec05201839da26fb7ea746cd15e23088c7e34a74","subject":"Merge \"Use opendev repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/ec05201839da26fb7ea746cd15e23088c7e34a74"}]}],"author":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-09-12 14:02:05.000000000","tz":-240},"committer":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-09-12 14:02:46.000000000","tz":-240},"subject":"Enable support for both python3.6 and python3.7","message":"Enable support for both python3.6 and python3.7\n\nChange-Id: Ia74d470a6d593a6a81779788e2848c5b5995c44f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/d27e32107047e18f388046b1ff627e894a9b0304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/d27e32107047e18f388046b1ff627e894a9b0304"}]},"branch":"refs/heads/master"},"7ca211e7007b4444b543457c8627c29ab37483cb":{"kind":"REWORK","_number":2,"created":"2019-09-17 03:52:22.000000000","uploader":{"_account_id":28063,"name":"VEDA","email":"va00121481@techmahindra.com","username":"VedaAnnayappa"},"ref":"refs/changes/55/681755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/55/681755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/55/681755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/55/681755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/55/681755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/55/681755/2"}}},"commit":{"parents":[{"commit":"f3008d1ee22329e0cd8e76aaf8b7b4a2a56af876","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/f3008d1ee22329e0cd8e76aaf8b7b4a2a56af876"}]}],"author":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-09-12 14:02:05.000000000","tz":-240},"committer":{"name":"VedaAnnayappa","email":"VA00121481@techmahindra.com","date":"2019-09-17 03:52:14.000000000","tz":-240},"subject":"Enable support for both python3.6 and python3.7","message":"Enable support for both python3.6 and python3.7\n\nChange-Id: Ia74d470a6d593a6a81779788e2848c5b5995c44f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/7ca211e7007b4444b543457c8627c29ab37483cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/7ca211e7007b4444b543457c8627c29ab37483cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
