)]}'
{"id":"openstack%2Fceilometer-powervm~691553","triplet_id":"openstack%2Fceilometer-powervm~master~I3b4ab8af6e13174bf80ccd809fd76a220a9fc7ad","project":"openstack/ceilometer-powervm","branch":"master","hashtags":[],"change_id":"I3b4ab8af6e13174bf80ccd809fd76a220a9fc7ad","subject":"Stop testing python2.7","status":"NEW","created":"2019-10-28 02:22:49.000000000","updated":"2020-05-19 01:04:26.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76cfb938a98d3f925e2f75c3add2bf2e3a979d6d","_number":691553,"virtual_id_number":691553,"owner":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-19 01:04:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-31 01:18:29.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-05-13 10:21:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-19 01:04:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55e961392e2eb5eadee2a21cb7bfbb0f12d8054a","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-28 02:22:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d6f1da9496f8878944fe1e9785018ede4f308d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 02:28:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844b883d8ddd45a1972dc6b60edfab41 : SUCCESS in 3m 39s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f8bac1e5aed4356a987cc6ee8d01dc1 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4de8a378d2f4e88b929c242f6bd2321 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f47687a54f13401b973cc55821c9b867 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/444960d30ff34934bf826c300fdd9b2c : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2d45e4da4184395bd04690f75e083bf : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eba555210cb74ce1a3f9ad79667923bc : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9f060bdba7275141eb6db7fe1308cb9391203c64","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-28 14:23:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d37355744c0a3027336c3183d61280c0a21aa155","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-29 00:45:58.000000000","message":"Patch Set 1:\n\n\u003e (5 comments)\n\nok，thanks very much！","accounts_in_message":[],"_revision_number":1},{"id":"67f40eddf16ad67e1bf23ec0eb2b85aa75543812","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-29 00:46:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e4919b922981b7ab3bb4f08d8e6276265036824","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-29 00:54:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c51bdf719c54c6a8d44c6c9e21339b3 : SUCCESS in 4m 06s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c78b705107b34b1ebeb4156229c23f7f : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7424e8cee21c48d7846f1e3fdeeda548 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15614f6b840845e58d116a08640b69e6 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f553f744d1af4ccd96d0cd19b33b838f : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e603c5d1f9f54914bf9c0876ebd842d5 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ada3e4b1414b1a3adaaab66dace94cf9e14e0fbe","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-30 18:35:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nOther than the basepython conflict warning (my fault!) this is doing the right thing -- all the zuul and tox jobs are running under py3 now.","accounts_in_message":[],"_revision_number":2},{"id":"fc42850d02862e40933dc34e500d0ad638adee60","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-31 00:37:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b2b628b23943b583559eee521824a6711efb90ab","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2019-10-31 00:38:22.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Other than the basepython conflict warning (my fault!) this is\n \u003e doing the right thing -- all the zuul and tox jobs are running\n \u003e under py3 now.\n\nI got it， please check again， Thanks very much","accounts_in_message":[],"_revision_number":2},{"id":"050f5047a33f83d7f87906a70e230c239eab5e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 00:47:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe9803b6c6b4b7a828d9922c953b981 : SUCCESS in 5m 18s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f11894bd998420a8973f661f6a58b6e : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4f67847a064158ab8b791e24c7ca7b : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/226811b4e7c248a5b25bf6f230c15cfc : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d37efc696f9405f81ba504158706de5 : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b4dc943c11748de898ab516f9789987 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":3},{"id":"fc8bcb24820c01184a96ce95e776e6ad558fa031","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-31 01:18:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6553f07ce859c569b203f557d07be2cfdebe7f94","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-13 10:21:08.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"06ce669d7d7cc8e5e85e663afeadc3af7cf16996","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-13 10:21:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16d0988b638477513bdba1e296441a18779ac87c","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-05-19 00:56:03.000000000","message":"Patch Set 3:\n\n\u003e (1 comment)\n\nok， i will update it，thanks","accounts_in_message":[],"_revision_number":3},{"id":"dfd5d0d63b6512bf7a76e78cbdf0d62a96e997fd","author":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"date":"2020-05-19 00:56:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"76cfb938a98d3f925e2f75c3add2bf2e3a979d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 01:04:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1ceec30fd54a8dbfb5754604c36edc : SUCCESS in 4m 52s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/238864c3affd4b46b1f98a8cdd41ff49 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560ac7691dc3495bab2b7a03733e230a : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7cd5ec8a4c14662ba90d43226d0e28f : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9791297bb004c02bd74792dcb000c2a : SUCCESS in 5m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b241ff13779d4b1f9e4da1fb3dc6a23b : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"67b7ebe3fee3df0ee8f14a74725922570b6f0d64","revisions":{"03eb30d6dde323265a0b4b8d448e76135f4c5560":{"kind":"REWORK","_number":1,"created":"2019-10-28 02:22:49.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/53/691553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/53/691553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/1"}}},"commit":{"parents":[{"commit":"7ca211e7007b4444b543457c8627c29ab37483cb","subject":"Enable support for both python3.6 and python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/7ca211e7007b4444b543457c8627c29ab37483cb"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 02:21:49.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 02:22:30.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: I3b4ab8af6e13174bf80ccd809fd76a220a9fc7ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/03eb30d6dde323265a0b4b8d448e76135f4c5560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/03eb30d6dde323265a0b4b8d448e76135f4c5560"}]},"branch":"refs/heads/master"},"2fa2a3973d7740ae56d4eed0b970f5d9fbd0fd87":{"kind":"REWORK","_number":2,"created":"2019-10-29 00:46:34.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/53/691553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/53/691553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/2"}}},"commit":{"parents":[{"commit":"7ca211e7007b4444b543457c8627c29ab37483cb","subject":"Enable support for both python3.6 and python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/7ca211e7007b4444b543457c8627c29ab37483cb"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 02:21:49.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-29 00:46:20.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: I3b4ab8af6e13174bf80ccd809fd76a220a9fc7ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/2fa2a3973d7740ae56d4eed0b970f5d9fbd0fd87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/2fa2a3973d7740ae56d4eed0b970f5d9fbd0fd87"}]},"branch":"refs/heads/master"},"555d328d4a45d919439767ac968324cc894098e2":{"kind":"REWORK","_number":3,"created":"2019-10-31 00:37:25.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/53/691553/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/53/691553/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/3"}}},"commit":{"parents":[{"commit":"7ca211e7007b4444b543457c8627c29ab37483cb","subject":"Enable support for both python3.6 and python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/7ca211e7007b4444b543457c8627c29ab37483cb"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 02:21:49.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-31 00:37:09.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: I3b4ab8af6e13174bf80ccd809fd76a220a9fc7ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/555d328d4a45d919439767ac968324cc894098e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/555d328d4a45d919439767ac968324cc894098e2"}]},"branch":"refs/heads/master"},"67b7ebe3fee3df0ee8f14a74725922570b6f0d64":{"kind":"REWORK","_number":4,"created":"2020-05-19 00:56:53.000000000","uploader":{"_account_id":30092,"name":"Xuan Yandong","email":"xuanyd@outlook.com","username":"xuanyandong"},"ref":"refs/changes/53/691553/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/53/691553/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/53/691553/4"}}},"commit":{"parents":[{"commit":"7ca211e7007b4444b543457c8627c29ab37483cb","subject":"Enable support for both python3.6 and python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/7ca211e7007b4444b543457c8627c29ab37483cb"}]}],"author":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2019-10-28 02:21:49.000000000","tz":480},"committer":{"name":"xuanyandong","email":"xuanyandong@inspur.com","date":"2020-05-19 00:56:38.000000000","tz":480},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: I3b4ab8af6e13174bf80ccd809fd76a220a9fc7ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/67b7ebe3fee3df0ee8f14a74725922570b6f0d64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/67b7ebe3fee3df0ee8f14a74725922570b6f0d64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
