)]}'
{"id":"openstack%2Fceilometer-powervm~734228","triplet_id":"openstack%2Fceilometer-powervm~master~I5012b09da389bc0363b9cbbce48ccf26d5be5215","project":"openstack/ceilometer-powervm","branch":"master","topic":"drop_mock","hashtags":[],"change_id":"I5012b09da389bc0363b9cbbce48ccf26d5be5215","subject":"Use unittest.mock instead of mock","status":"NEW","created":"2020-06-08 20:26:40.000000000","updated":"2020-06-08 22:04:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21930ba5852a1d7497d685d180fbeacdfc8cafe6","_number":734228,"virtual_id_number":734228,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-08 20:53:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2020-06-08 22:04:25.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 20:53:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2fa4a7f1311addc27f20fd66c082b7437cd82eda","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 20:26:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b56c467f1a20b800317c28005b431d5ef930624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:53:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355225ad46c84abf95f0ed4cbfeb137d : SUCCESS in 7m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8384fd3c1c8141c2976c7e756936a289 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665662ff61a94ce4849c4b5e8dd02692 : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2e80cff9c9a464989d233ad891fb3a0 : FAILURE in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40a26833babe4b488e4eb9c8479250f6 : FAILURE in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6deba57a82794a4582a1bce19eace258 : FAILURE in 7m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75a980fa6f154b3287b447ab0281ec56 : SUCCESS in 9m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ec7728a398ae4420113dbd38d704baf8fa547894","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-08 22:04:25.000000000","message":"Patch Set 1: Workflow-1\n\nStill support py27","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71da01d16a3a4ceab5411dd669c1bd6951288162","revisions":{"71da01d16a3a4ceab5411dd669c1bd6951288162":{"kind":"REWORK","_number":1,"created":"2020-06-08 20:26:40.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/28/734228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer-powervm","ref":"refs/changes/28/734228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/28/734228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/28/734228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer-powervm refs/changes/28/734228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer-powervm refs/changes/28/734228/1"}}},"commit":{"parents":[{"commit":"7ca211e7007b4444b543457c8627c29ab37483cb","subject":"Enable support for both python3.6 and python3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/7ca211e7007b4444b543457c8627c29ab37483cb"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 20:25:54.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-08 20:25:54.000000000","tz":120},"subject":"Use unittest.mock instead of mock","message":"Use unittest.mock instead of mock\n\nThe mock third party library was needed for mock support in py2\nruntimes. Since we now only support py36 and later, we can use the\nstandard lib unittest.mock module instead.\n\nChange-Id: I5012b09da389bc0363b9cbbce48ccf26d5be5215\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/71da01d16a3a4ceab5411dd669c1bd6951288162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer-powervm/commit/71da01d16a3a4ceab5411dd669c1bd6951288162"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
