)]}'
{"id":"openstack%2Fceilometermiddleware~175918","triplet_id":"openstack%2Fceilometermiddleware~master~I11b17f90ad351f369f012eb954da9f63804b4eb5","project":"openstack/ceilometermiddleware","branch":"master","topic":"update-readme","hashtags":[],"change_id":"I11b17f90ad351f369f012eb954da9f63804b4eb5","subject":"Update README to work with release tools","status":"MERGED","created":"2015-04-21 14:58:55.000000000","updated":"2015-04-22 14:37:57.000000000","submitted":"2015-04-22 14:37:55.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e0e8650da9bca575c7be7d958920badea88b41db","_number":175918,"virtual_id_number":175918,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2015-04-22 14:37:55.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-04-21 17:14:11.000000000","_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-04-22 13:42:54.000000000","_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-04-22 13:42:54.000000000","updated_by":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"reviewer":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"state":"REVIEWER"},{"updated":"2015-04-22 14:37:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"becab302b8e098f05015895f3df55439c8aafe1e","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2015-04-21 14:58:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df2c65b5272111fd91f1bdd9e7cebd272ee4c516","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-21 16:03:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-ceilometermiddleware-pep8 http://logs.openstack.org/18/175918/1/check/gate-ceilometermiddleware-pep8/6f851d9/ : SUCCESS in 53s\n- gate-ceilometermiddleware-docs http://docs-draft.openstack.org/18/175918/1/check/gate-ceilometermiddleware-docs/b8e3f42//doc/build/html/ : SUCCESS in 44s\n- gate-ceilometermiddleware-python27 http://logs.openstack.org/18/175918/1/check/gate-ceilometermiddleware-python27/0ec3d11/ : SUCCESS in 1m 01s\n- gate-ceilometermiddleware-python34 http://logs.openstack.org/18/175918/1/check/gate-ceilometermiddleware-python34/a48b41c/ : SUCCESS in 2m 45s\n- gate-tempest-dsvm-neutron-src-ceilometermiddleware http://logs.openstack.org/18/175918/1/check/gate-tempest-dsvm-neutron-src-ceilometermiddleware/044322b/ : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1d88ba2d7e1f01f677edb5daa958bbccef941b9e","author":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"date":"2015-04-21 17:14:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"586c8e254ab089fdc7691438e03d73d2bdd63df5","author":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"date":"2015-04-22 13:42:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aaf925098a2a35cace1337716807d4e20e9c0b7a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 13:42:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"07bbd71589941b65fc53f64fc55875bd7c34e31c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 14:37:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-ceilometermiddleware-docs http://docs-draft.openstack.org/18/175918/1/gate/gate-ceilometermiddleware-docs/09019f7//doc/build/html/ : SUCCESS in 45s\n- gate-ceilometermiddleware-pep8 http://logs.openstack.org/18/175918/1/gate/gate-ceilometermiddleware-pep8/7024d37/ : SUCCESS in 42s\n- gate-ceilometermiddleware-python27 http://logs.openstack.org/18/175918/1/gate/gate-ceilometermiddleware-python27/4b752f1/ : SUCCESS in 45s\n- gate-ceilometermiddleware-python34 http://logs.openstack.org/18/175918/1/gate/gate-ceilometermiddleware-python34/0e88243/ : SUCCESS in 2m 05s\n- gate-tempest-dsvm-neutron-src-ceilometermiddleware http://logs.openstack.org/18/175918/1/gate/gate-tempest-dsvm-neutron-src-ceilometermiddleware/48e5386/ : SUCCESS in 54m 32s","accounts_in_message":[],"_revision_number":1},{"id":"48d01e77f3be20c4acf210c36f449e792a3571f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-22 14:37:57.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fc5037cdfae209825c9122cb0851ff70538e698","revisions":{"1fc5037cdfae209825c9122cb0851ff70538e698":{"kind":"REWORK","_number":1,"created":"2015-04-21 14:58:55.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/18/175918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometermiddleware","ref":"refs/changes/18/175918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/18/175918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/18/175918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/18/175918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometermiddleware refs/changes/18/175918/1"}}},"commit":{"parents":[{"commit":"3a7aaa8787cf968f9ff3ef7c36aeeb592baf4b07","subject":"Merge \"use pycadf helper module to assign action\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/3a7aaa8787cf968f9ff3ef7c36aeeb592baf4b07"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2015-04-21 14:58:38.000000000","tz":0},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2015-04-21 14:58:42.000000000","tz":0},"subject":"Update README to work with release tools","message":"Update README to work with release tools\n\nThe README file needs to have links to the project documentation and bug\ntracker in a parsable format in order for some of the release tools\nscripts to work (particularly the one that prints the release note\nemail).\n\nChange-Id: I11b17f90ad351f369f012eb954da9f63804b4eb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/1fc5037cdfae209825c9122cb0851ff70538e698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/1fc5037cdfae209825c9122cb0851ff70538e698"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3a7aaa8787cf968f9ff3ef7c36aeeb592baf4b07","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
