)]}'
{"id":"openstack%2Fceilometermiddleware~719275","triplet_id":"openstack%2Fceilometermiddleware~master~I48909423d8b440c9bea096d78e9a43721a64a44e","project":"openstack/ceilometermiddleware","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"I48909423d8b440c9bea096d78e9a43721a64a44e","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-04-11 18:49:30.000000000","updated":"2020-04-13 01:41:54.000000000","submitted":"2020-04-13 01:39:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719275-1586741996164-76a6aaaf","meta_rev_id":"95c38ae3b9429776eb9d8a9aaf497032d5785180","_number":719275,"virtual_id_number":719275,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 01:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-04-13 00:54:29.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-04-13 00:54:29.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 00:54:29.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-04-13 01:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f15d4c16d92936f83bc1d478ab65e484b1905aad","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-11 18:49:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8962a865c71571bb841e8c367159741768c2c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 19:48:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515a3ca6d21f436c9393e14fc23896ba : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92975a9c833b4014bac1496fd7095aeb : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d2239ff7a634e56bb80153178fb7e28 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15cb1a7c07134496989467ae27fadc9e : SUCCESS in 4m 19s\n- ceilometermiddleware-integration https://zuul.opendev.org/t/openstack/build/277f309775784435ba50eaf9539ce267 : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":1},{"id":"161be39a56ca2a5e2ca181a6055474c1cb24a895","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 00:54:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f55c7e3bd1470dc1a412afd4b9c7662e3a2d43d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 00:55:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"46f81ac7835c4b184dee3125b85d4a81e7aaaa53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 01:39:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e36176252354405da69fa6b2949fa385 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aed7db7242fa418181675c49daf26f5a : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13bc2e7766e741e1830896bb09a1d80a : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9603a125977744619f3cd15e4c6fdfbd : SUCCESS in 3m 25s\n- ceilometermiddleware-integration https://zuul.opendev.org/t/openstack/build/e956d87601a9402fbf78cca88d373812 : SUCCESS in 43m 12s","accounts_in_message":[],"_revision_number":1},{"id":"65d479489e20a96eaba58aeb8b9addf732ca7e79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 01:39:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b884b0c1206401f1f91a6ef0c6cf7604d0caa869","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 01:41:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b007ec9123484d67b81f66f3fccfa072 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"deacd8d7821cdee03fa9c86409c90e046e349acd","revisions":{"deacd8d7821cdee03fa9c86409c90e046e349acd":{"kind":"REWORK","_number":1,"created":"2020-04-11 18:49:30.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/75/719275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometermiddleware","ref":"refs/changes/75/719275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/75/719275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/75/719275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/75/719275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometermiddleware refs/changes/75/719275/1"}}},"commit":{"parents":[{"commit":"e80e0f4c86188cd31905f3ad29e5d9c2c09a69f8","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/e80e0f4c86188cd31905f3ad29e5d9c2c09a69f8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-11 18:49:28.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-11 18:49:28.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I48909423d8b440c9bea096d78e9a43721a64a44e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/deacd8d7821cdee03fa9c86409c90e046e349acd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/deacd8d7821cdee03fa9c86409c90e046e349acd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
