)]}'
{"id":"openstack%2Fceilometermiddleware~906957","triplet_id":"openstack%2Fceilometermiddleware~master~I170c1b5e15229ab4cf043773208b3a76e3e02188","project":"openstack/ceilometermiddleware","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2024-02-21 11:20:36.000000000","reason":"Change was submitted"},"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-02-21 11:20:36.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-21 09:30:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-02-21 06:54:34.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"hashtags":[],"change_id":"I170c1b5e15229ab4cf043773208b3a76e3e02188","subject":"Bump hacking","status":"MERGED","created":"2024-01-27 13:25:57.000000000","updated":"2024-02-21 11:21:32.000000000","submitted":"2024-02-21 11:20:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"906957-bump-hacking","meta_rev_id":"b26fa361fd130c9e0a02aa6d40e4ea7ca84ffa5e","_number":906957,"virtual_id_number":906957,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 11:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"recommended":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"all":[{"value":1,"date":"2024-01-31 07:00:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":2,"date":"2024-02-21 06:54:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2024-01-30 15:42:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-02-21 06:54:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-27 14:21:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-30 15:42:30.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-01-31 07:00:06.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-02-02 12:41:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-02-02 12:41:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"REVIEWER"},{"updated":"2024-02-02 12:41:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"27d09c92a57928c33eee4e141bb7c8bc1eb56ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-27 13:25:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a0aa24ef5ae3d945947a1ed533bcc03f44b5e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 14:21:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b8595ab0f8343e39a0a257cd2b5f33f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c5932fc70964419bb0952600fb84e55 : SUCCESS in 2m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84dfcaf6d0d949b3a051bd8801c4a967 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18fb64231157467d90991c67467f457d : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8452792bfc8c45148f1c01f464517e78 : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d81104ce3c574541ac5325e6ad6fd1dc : SUCCESS in 2m 44s\n- ceilometermiddleware-integration https://zuul.opendev.org/t/openstack/build/77cc2f990b124da58c9fcfcda1332f91 : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":1},{"id":"de89679f982d0c489d0bc5cde35ef40837cff95a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-30 15:42:30.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79067a52ce6d93a7b193c7819d517db68be90a32","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-01-31 07:00:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"007b5853ae91fd6a13442814fa446792b56216d6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 04:21:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fcf33dc26b3d62f6883846fc508270462a41d51a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-02-21 06:54:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"132a4032123c7b20565185c77fd148a35bdf96c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 06:55:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14db6fb72a538f79b3da9cd8f67ed0c06cbcf38c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 07:38:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72f8102b23334688bcaed93dc07ebe8e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a218b92a79af44e59100d35b2ac514fb : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2b42353cf31438782570a92fc7ebd7a : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8afaaf48e3c4c349fdb354100891f80 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7d74098187a429abda1aaa4d3395382 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eea2588b5e2e4bd0bba5cead016eee37 : FAILURE in 2m 28s\n- ceilometermiddleware-integration https://zuul.opendev.org/t/openstack/build/ee1870293d0447ed8bc7d731fce7b060 : SUCCESS in 42m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9cf6f3af661c2fa5a9df5131caabea4546e63bbc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-21 09:30:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de28b7290a9c07aecce81b0f2d18126dce6a94dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 10:31:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1d8795682ed4afa82bc08dc12757dee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44773fcc9f4e49d58b4e11291ac2dbc4 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b51fe69006a4777bc6dfbbd9fbd4dae : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee14ea10f250423d97f6cf45b242d236 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a01b17b5f9444ebeb717b4df9b3dc4 : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b24164837a14660bf1c19e545e8d9f0 : SUCCESS in 2m 35s\n- ceilometermiddleware-integration https://zuul.opendev.org/t/openstack/build/76fb66b6fed04ad8b5b3a0fe64754c3f : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a5b0b9590883abefc78758a99638a4ce54484a0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 10:31:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1eab72d2f122028700ea5fa88061a40741e53734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 11:20:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bad73edba7c94437a07b974ef59f1e20\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62966907d1d4b918b45857d3a8e28fb : SUCCESS in 1m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83751feb981a4fab89a92c3c422e745c : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9066618cafb34b5bb092662db0005c24 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f648d36a3e48dba16328120503fa7a : SUCCESS in 2m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cd2af2322db4ad599af6490f8d79d6e : SUCCESS in 2m 37s\n- ceilometermiddleware-integration https://zuul.opendev.org/t/openstack/build/59d5ce83c8e7432ba96cc8fff03c176a : SUCCESS in 47m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a46d5ac8fb01d091b32900f6daa0512cb743254e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 11:20:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b26fa361fd130c9e0a02aa6d40e4ea7ca84ffa5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 11:21:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daaa3e37dc944551a594cdcdc4bc8ba0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7e8e2fec204027ab19e93e98afe115 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a6a905681b44687a50d4be85f4af69d : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce3dee35635085f59947f967467cf6bffac631e9","revisions":{"ce3dee35635085f59947f967467cf6bffac631e9":{"kind":"REWORK","_number":1,"created":"2024-01-27 13:25:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/906957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometermiddleware","ref":"refs/changes/57/906957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/57/906957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/57/906957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometermiddleware refs/changes/57/906957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometermiddleware refs/changes/57/906957/1"}}},"commit":{"parents":[{"commit":"0689769ba847f9222ce6b7c92e6bc3757e5d608a","subject":"Ignore .eggs directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/0689769ba847f9222ce6b7c92e6bc3757e5d608a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 13:25:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-27 13:25:42.000000000","tz":540},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\n\nChange-Id: I170c1b5e15229ab4cf043773208b3a76e3e02188\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/ce3dee35635085f59947f967467cf6bffac631e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometermiddleware/commit/ce3dee35635085f59947f967467cf6bffac631e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
