)]}'
{"id":"openstack%2Fceilometer~1004553","triplet_id":"openstack%2Fceilometer~master~I220e2b1a464a4620e4b2cdd8226ea1ffab638c96","project":"openstack/ceilometer","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2026-09-08 07:36:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I220e2b1a464a4620e4b2cdd8226ea1ffab638c96","subject":"Fix typo in warning message","status":"MERGED","created":"2026-09-07 22:10:30.000000000","updated":"2026-09-08 07:37:48.000000000","submitted":"2026-09-08 07:36:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004553","meta_rev_id":"8fb09cdcd90eec056731b9cccb0dfe890aed97f5","_number":1004553,"virtual_id_number":1004553,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-08 07:36:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2026-09-07 23:21:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-09-08 06:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-09-08 06:39:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 23:21:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-07 23:21:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-09-08 06:39:52.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"53ff4926811ab79a07d401c4b504c5f9da0a0d29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-09-07 22:10:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"735deb49fe08eaee44fbf2a4fec9226adf0b93ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:21:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4575c7ccdaaf4b9d915cef368f34b8be\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd62c4fc101747f79e990e93bc742f3d : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bd4c43fb70b477fb266a6b502913e7e : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b22a491888f949b6aedb75ece2c9c648 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f259735aea744cf8968ca10fb4fc075f : SUCCESS in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc0643b1c76a424b84047e0ef33344a1 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f27be238dbb64577b5245151cc6a52d9 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f285b64369540dd9b3b53bebfc1a892 : SUCCESS in 2m 52s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/54d858474ff049acbf414ac03645f17d : SUCCESS in 35m 50s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/6ec466ccaebf43d8b21cee8cf0f0a830 : SUCCESS in 1h 00m 12s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/b5fb552ece314d209bc741cc13f6813e : SUCCESS in 1h 03m 51s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/529e8bfb404d45379835c294923146f4 : SUCCESS in 33m 01s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/a42b549939fb4907838090c8cf370270 : SUCCESS in 41m 51s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6d3180c9049d40e6ad2171019378392d : SUCCESS in 52m 43s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/80618b43a414473ab6134f4768fdd520 : SUCCESS in 42m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"f93c322b67d3a47ebefd458e6cc4a937d69b70d6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 23:21:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc8e1c1ae010533129f460702cceac4a691258f1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-09-08 06:39:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e8e8e8e7333d485721d3144ec11cc008c6bbcfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 06:40:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2bce28c2e44d2d7a167a78f0cbf9b83fe70be06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 07:36:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b0fbb240b2d4a09b199805c55aecb6c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee86f2f201a4d9384ee44538889abf2 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/304d94cd322e41b3b27096407eda9f57 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/279a1a57715746d6815d344a23344e96 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7d864e99b14e8da7e7a09f163cc628 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c58a3c57df4436396755bdc59312335 : SUCCESS in 6m 11s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/9b1bca2d3f3d48b99f52e1d0f4ca0191 : SUCCESS in 50m 46s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3d2b5e57150443dda99c2209d0b73248 : SUCCESS in 48m 30s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/ec1177f8fabe46aead53b1a66d8549f5 : SUCCESS in 48m 13s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f847c90c656640238c1a93d7187c0dca : SUCCESS in 49m 23s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/fbea42170d654b33a7d0a7561b90794c : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6bc6d34239b8caa48a93b7985fd975f8d9e6ab80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 07:36:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8fb09cdcd90eec056731b9cccb0dfe890aed97f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-08 07:37:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c14036712c044a3bcbf23e194c0715d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30e17d6c37442baa52698793c7a39a3 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae6056d501ec669f426b4cc1a3861b326eb8c5ab","revisions":{"ae6056d501ec669f426b4cc1a3861b326eb8c5ab":{"kind":"REWORK","_number":1,"created":"2026-09-07 22:10:30.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/53/1004553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/53/1004553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/53/1004553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/53/1004553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/53/1004553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/53/1004553/1"}}},"commit":{"parents":[{"commit":"9a6ea17568a062816261629db57c5405a93518a5","subject":"Merge \"Remove deprecated os_cacert options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/9a6ea17568a062816261629db57c5405a93518a5"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-09-07 22:09:54.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2026-09-07 22:09:54.000000000","tz":120},"subject":"Fix typo in warning message","message":"Fix typo in warning message\n\n... this was wrong when it was introduced and\nshould be interfaceStats not instanceStats.\n\nChange-Id: I220e2b1a464a4620e4b2cdd8226ea1ffab638c96\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/ae6056d501ec669f426b4cc1a3861b326eb8c5ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/ae6056d501ec669f426b4cc1a3861b326eb8c5ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
