)]}'
{"id":"openstack%2Fceilometer~101799","triplet_id":"openstack%2Fceilometer~stable%2Fhavana~Ia1bfa1bd24989681db1d2f385defc12e69a01f8d","project":"openstack/ceilometer","branch":"stable/havana","topic":"bug/1321080/oslo-incubator-sync","hashtags":[],"change_id":"Ia1bfa1bd24989681db1d2f385defc12e69a01f8d","subject":"remove token from notifier middleware","status":"MERGED","created":"2014-06-23 05:21:37.000000000","updated":"2014-06-23 14:57:07.000000000","submitted":"2014-06-23 14:57:06.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac34b5a6a013b885b8a51dfde6bc17126abc5ca8","_number":101799,"virtual_id_number":101799,"owner":{"_account_id":7473,"name":"Grant Murphy","email":"grantcmurphy@gmail.com","username":"gm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2014-06-23 14:57:06.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-06-23 14:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-06-23 14:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-06-23 14:14:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2014-06-23 14:57:06.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"c0f440c754b816e7f7e276afa30b784c49b6a34d","author":{"_account_id":7473,"name":"Grant Murphy","email":"grantcmurphy@gmail.com","username":"gm"},"date":"2014-06-23 05:21:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a089e74d1bdbda979202c4070b9d95a9d3dd8b07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-23 05:21:38.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"89c9edc549ada6093f9f077fb04f6e4b8329c834","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-23 06:03:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-ceilometer-pep8 http://logs.openstack.org/99/101799/1/check/gate-ceilometer-pep8/20985e7 : SUCCESS in 4m 08s\n- gate-ceilometer-docs http://docs-draft.openstack.org/99/101799/1/check/gate-ceilometer-docs/fc95887/doc/build/html/ : SUCCESS in 8m 44s\n- gate-ceilometer-python26 http://logs.openstack.org/99/101799/1/check/gate-ceilometer-python26/40ecc3a : SUCCESS in 8m 28s\n- gate-ceilometer-python27 http://logs.openstack.org/99/101799/1/check/gate-ceilometer-python27/831e22b : SUCCESS in 9m 08s\n- check-tempest-dsvm-full http://logs.openstack.org/99/101799/1/check/check-tempest-dsvm-full/5538e13 : SUCCESS in 37m 05s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/99/101799/1/check/check-tempest-dsvm-postgres-full/1d8df46 : SUCCESS in 34m 31s\n- check-tempest-dsvm-neutron http://logs.openstack.org/99/101799/1/check/check-tempest-dsvm-neutron/12984ba : SUCCESS in 34m 01s\n- check-devstack-dsvm-cells http://logs.openstack.org/99/101799/1/check/check-devstack-dsvm-cells/188ac5c : SUCCESS in 12m 37s\n- check-swift-dsvm-functional http://logs.openstack.org/99/101799/1/check/check-swift-dsvm-functional/58397b0 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a87aa1228fcc4b7cbb0d7c36660d2e89a51ba3da","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2014-06-23 14:14:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSecurity fix","accounts_in_message":[],"_revision_number":1},{"id":"0bdf958e89bf806230dd28f8e4f43b4dc20e1579","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-23 14:14:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"71e8f848afa002ab65af5ae4690632f5b917d4b8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-23 14:57:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded.\n\n- gate-ceilometer-docs http://docs-draft.openstack.org/99/101799/1/gate/gate-ceilometer-docs/c8479d1/doc/build/html/ : SUCCESS in 6m 43s\n- gate-ceilometer-pep8 http://logs.openstack.org/99/101799/1/gate/gate-ceilometer-pep8/a942f26 : SUCCESS in 4m 22s\n- gate-ceilometer-python26 http://logs.openstack.org/99/101799/1/gate/gate-ceilometer-python26/06aa0e3 : SUCCESS in 9m 25s\n- gate-ceilometer-python27 http://logs.openstack.org/99/101799/1/gate/gate-ceilometer-python27/a87c03e : SUCCESS in 10m 51s\n- gate-tempest-dsvm-full http://logs.openstack.org/99/101799/1/gate/gate-tempest-dsvm-full/5d5b431 : SUCCESS in 37m 05s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/99/101799/1/gate/gate-tempest-dsvm-postgres-full/3b4c68d : SUCCESS in 36m 04s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/99/101799/1/gate/gate-tempest-dsvm-neutron/ca1cf97 : SUCCESS in 35m 08s\n- gate-devstack-dsvm-cells http://logs.openstack.org/99/101799/1/gate/gate-devstack-dsvm-cells/6e90bcf : SUCCESS in 14m 05s\n- gate-swift-dsvm-functional http://logs.openstack.org/99/101799/1/gate/gate-swift-dsvm-functional/4d0e01a : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1fc8a9d4a069715ff3f8e0f3d70e5f3391113a33","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-23 14:57:07.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"264f3b0d9640edeac743f339786e0a3b22c0f6c2","revisions":{"264f3b0d9640edeac743f339786e0a3b22c0f6c2":{"kind":"REWORK","_number":1,"created":"2014-06-23 05:21:37.000000000","uploader":{"_account_id":7473,"name":"Grant Murphy","email":"grantcmurphy@gmail.com","username":"gm"},"ref":"refs/changes/99/101799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/99/101799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/99/101799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/99/101799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/99/101799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/99/101799/1"}}},"commit":{"parents":[{"commit":"a2c909ca4bb32d4a82e2658a6e1faa972f9fea12","subject":"Fix testing gate due to new keystoneclient release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/a2c909ca4bb32d4a82e2658a6e1faa972f9fea12"}]}],"author":{"name":"Grant Murphy","email":"gmurphy@redhat.com","date":"2014-06-23 05:07:54.000000000","tz":0},"committer":{"name":"Grant Murphy","email":"gmurphy@redhat.com","date":"2014-06-23 05:07:54.000000000","tz":0},"subject":"remove token from notifier middleware","message":"remove token from notifier middleware\n\noslo-incubator sync to address the security bug\nin middleware (as below).\n\nnotifier middleware is capturing token and sending it to MQ. this\nis not advisable so we should filter it out.\n\nChange-Id: Ia1bfa1bd24989681db1d2f385defc12e69a01f8d\nCloses-Bug: #1321080\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/264f3b0d9640edeac743f339786e0a3b22c0f6c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/264f3b0d9640edeac743f339786e0a3b22c0f6c2"}]},"branch":"refs/heads/stable/havana"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
