)]}'
{"id":"openstack%2Fceilometer~614400","triplet_id":"openstack%2Fceilometer~master~I6e76b74b2f02251ee39025e02de44b13568ebed2","project":"openstack/ceilometer","branch":"master","topic":"upgrade-checkers","hashtags":[],"change_id":"I6e76b74b2f02251ee39025e02de44b13568ebed2","subject":"Add framework for ceilometer-status upgrade check","status":"MERGED","created":"2018-10-31 04:20:41.000000000","updated":"2019-05-17 05:53:44.000000000","submitted":"2019-05-16 06:46:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"614400-1557989173850-7c1c31b6","meta_rev_id":"b6f878e8ac88ef348df4a856b0952b2f079e5b0b","_number":614400,"virtual_id_number":614400,"owner":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},{"value":0,"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-05-16 06:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"recommended":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"all":[{"value":0,"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},{"value":0,"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":1,"date":"2019-05-16 04:43:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-05-16 05:09:08.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":2,"date":"2019-05-16 05:39:23.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},{"value":0,"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-05-16 05:39:23.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-07 01:33:02.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2018-11-19 04:52:38.000000000","updated_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"reviewer":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"state":"REVIEWER"},{"updated":"2018-11-19 04:52:55.000000000","updated_by":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"reviewer":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"state":"REVIEWER"},{"updated":"2018-12-05 20:55:10.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2018-12-05 21:30:32.000000000","updated_by":{"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},"reviewer":{"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},"state":"REVIEWER"},{"updated":"2019-01-08 05:10:05.000000000","updated_by":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"reviewer":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"state":"REVIEWER"},{"updated":"2019-03-16 09:40:42.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-03-18 05:59:55.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-05-16 04:43:09.000000000","updated_by":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"reviewer":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"state":"REVIEWER"},{"updated":"2019-05-16 05:09:08.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-05-16 05:39:23.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-05-16 06:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d877f3dcf6fcf1d5be694a39a69110bf9c10a1a","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2018-10-31 04:20:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65d6bd73473bff9dc143221d722e982e04ae2765","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2018-10-31 04:26:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3305afecd11137b0cb62f3259a4ab8b29fc7b2f9","author":{"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},"date":"2018-10-31 04:31:58.000000000","message":"Patch Set 1:\n\n* ipmi-test http://52.27.155.124/ipmi/614400/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d227a5164ed928ba036278c9726bcdd5c45a00b7","author":{"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},"date":"2018-10-31 04:42:51.000000000","message":"Patch Set 2:\n\n* ipmi-test http://52.27.155.124/ipmi/614400/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"874c9053757d9fea89d8f5bc2da8a13ccbe42da7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-31 05:38:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack-python3 http://logs.openstack.org/00/614400/2/check/test-release-openstack-python3/a92d622/ : SUCCESS in 2m 35s\n- openstack-tox-cover http://logs.openstack.org/00/614400/2/check/openstack-tox-cover/16d14fb/cover/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/00/614400/2/check/openstack-tox-pep8/58bc939/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/00/614400/2/check/openstack-tox-py27/4d3413b/ : SUCCESS in 4m 24s\n- openstack-tox-py35 http://logs.openstack.org/00/614400/2/check/openstack-tox-py35/c1d80b1/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/00/614400/2/check/openstack-tox-py36/fb007f2/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/00/614400/2/check/openstack-tox-docs/7bdeea9/html/ : SUCCESS in 3m 05s\n- build-openstack-releasenotes http://logs.openstack.org/00/614400/2/check/build-openstack-releasenotes/b799ad5/html/ : SUCCESS in 4m 00s\n- telemetry-tox-py37 http://logs.openstack.org/00/614400/2/check/telemetry-tox-py37/ba87e2c/ : SUCCESS in 5m 59s\n- grenade-dsvm-ceilometer http://logs.openstack.org/00/614400/2/check/grenade-dsvm-ceilometer/5bd3a46/ : SUCCESS in 1h 09m 29s\n- telemetry-dsvm-integration http://logs.openstack.org/00/614400/2/check/telemetry-dsvm-integration/8a14cd8/ : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":2},{"id":"07ded7543268d9a24a7dc5e635579d72762e413a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2018-10-31 15:20:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"04a3c3f5f6d58e41fb32364c8cea83087668735c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-11-07 01:33:02.000000000","message":"Patch Set 2: Code-Review+1\n\nlgtm, no lower_contraints.txt here.","accounts_in_message":[],"_revision_number":2},{"id":"5a509d65ae8d727e424352757cfa5770c866024b","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-11-12 23:45:47.000000000","message":"Patch Set 2: Code-Review+1\n\nThis is a pretty standard first implementation across all our OpenStack projects (that haven\u0027t implemented real checks already).","accounts_in_message":[],"_revision_number":2},{"id":"baf635cdf6bf8d9cd762d27497c17f8d36804cfe","author":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"date":"2018-12-04 04:52:16.000000000","message":"Removed reviewer Rajat Dhasmana with the following votes:\n\n* Code-Review+1 by Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"2931734af8b5c4d3b711c943b40eb5168b44028f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-12-05 20:54:40.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fb8e9b43c474172ae3825dff31e7648204000464","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-12-05 20:54:53.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"8b63c9db4ec6fe1ddc7ef79ee956937aef97b3b0","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2018-12-05 20:55:10.000000000","message":"Patch Set 3: Code-Review+1\n\nUpdated the required version of oslo.upgradecheck to 0.1.1. Note there is no lower-constraints yet in ceilometer.","accounts_in_message":[],"_revision_number":3},{"id":"0d492be69f0355604f9956488a0863a5a7273920","author":{"_account_id":15843,"name":"Intel Hardware-Meters CI","email":"ipmi-ci@intel.com","username":"ipmici"},"date":"2018-12-05 21:30:32.000000000","message":"Patch Set 3:\n\n* ipmi-test http://52.27.155.124/ipmi/614400/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"2ba9514abdc025cbf9a401d1bcea5ebde71e3bf1","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2018-12-05 22:03:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"06648974245ebfb1ba33e82568220a53993efc4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-05 22:25:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/00/614400/3/check/test-release-openstack/3c2c1b3/ : SUCCESS in 2m 58s\n- openstack-tox-cover http://logs.openstack.org/00/614400/3/check/openstack-tox-cover/2ebac41/cover/ : SUCCESS in 6m 28s\n- openstack-tox-pep8 http://logs.openstack.org/00/614400/3/check/openstack-tox-pep8/499b9cf/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/00/614400/3/check/openstack-tox-py27/93f3cf9/ : SUCCESS in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/00/614400/3/check/openstack-tox-py35/e4088ff/ : SUCCESS in 6m 24s\n- openstack-tox-py36 http://logs.openstack.org/00/614400/3/check/openstack-tox-py36/11a0f7a/ : SUCCESS in 5m 52s\n- openstack-tox-docs http://logs.openstack.org/00/614400/3/check/openstack-tox-docs/ec09284/html/ : SUCCESS in 3m 28s\n- build-openstack-releasenotes http://logs.openstack.org/00/614400/3/check/build-openstack-releasenotes/e764e95/html/ : SUCCESS in 6m 14s\n- telemetry-tox-py37 http://logs.openstack.org/00/614400/3/check/telemetry-tox-py37/d2169d5/ : SUCCESS in 6m 26s\n- grenade-dsvm-ceilometer http://logs.openstack.org/00/614400/3/check/grenade-dsvm-ceilometer/9addfce/ : SUCCESS in 56m 12s\n- telemetry-dsvm-integration http://logs.openstack.org/00/614400/3/check/telemetry-dsvm-integration/21a2c79/ : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":3},{"id":"6ecb03703683ee9a3d294494d9414945bf2fb21e","author":{"_account_id":29360,"name":"Prankul Mahajan","email":"prankul.mahajan88@gmail.com","username":"prankul"},"date":"2019-01-08 05:10:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0c91eddc1e45197efd8347120dba55b79a627ce7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-03-16 09:40:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ac10809184ef2b42644b59300d0bd6023855f5cd","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-03-18 05:59:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"21165489825f0f30526b625ce3bd5abc0738f876","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-14 23:21:26.000000000","message":"Patch Set 3:\n\nShould we rebase this on master and see if we can get past the \"cannot merge\" error?","accounts_in_message":[],"_revision_number":3},{"id":"91679cd8d0302e707b6fd3ef9a5a866021700b20","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-15 02:04:37.000000000","message":"Patch Set 3:\n\nHi Joseph. That would be great. Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"96fec97d63d1df0036a856c2b1e1ed207d35792a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-05-16 03:15:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"588a98e7879e457ccc60f0b2353cc510df2be93a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-05-16 03:16:17.000000000","message":"Patch Set 4:\n\njust rebased this patch","accounts_in_message":[],"_revision_number":4},{"id":"be8edeada95b9f046946f33b8c97a70e5a34e633","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 04:11:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/00/614400/4/check/test-release-openstack/3252e53/ : SUCCESS in 2m 23s\n- openstack-tox-cover http://logs.openstack.org/00/614400/4/check/openstack-tox-cover/64759b0/cover/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/00/614400/4/check/openstack-tox-pep8/ff22aaf/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/00/614400/4/check/openstack-tox-py27/9d3b254/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/00/614400/4/check/openstack-tox-py36/9b48dcc/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/00/614400/4/check/openstack-tox-py37/23f509f/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/00/614400/4/check/openstack-tox-docs/3ca9e2b/html/ : SUCCESS in 4m 23s\n- build-openstack-releasenotes http://logs.openstack.org/00/614400/4/check/build-openstack-releasenotes/f244d88/html/ : SUCCESS in 5m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/614400/4/check/openstack-tox-lower-constraints/696f1c2/ : SUCCESS in 6m 12s\n- grenade-dsvm-ceilometer http://logs.openstack.org/00/614400/4/check/grenade-dsvm-ceilometer/4bb9318/ : SUCCESS in 52m 45s\n- telemetry-dsvm-integration http://logs.openstack.org/00/614400/4/check/telemetry-dsvm-integration/3813b2e/ : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":4},{"id":"bf9a98899a9dddb2f5b1fe8b25e696ee81308517","author":{"_account_id":10311,"name":"Joseph Davis","email":"joseph.davis@suse.com","username":"joadavis"},"date":"2019-05-16 04:43:09.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"48d0b57e88f9ff4cc53f9a55a224370dd7565e1d","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-05-16 05:09:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f6a6549ebe2d80b624ef61eec12cf1bc90a7e071","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-16 05:39:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":4},{"id":"707c1def6e40c1ca53991dc80861ea9d8142ea8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 05:39:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0ecffa3ac064758ada3150134221dbc19aa7d1ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:46:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"fb22414935d8c39c3f7281ae937673aff5c16e43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:46:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/00/614400/4/gate/test-release-openstack/b6c9352/ : SUCCESS in 1m 58s\n- openstack-tox-pep8 http://logs.openstack.org/00/614400/4/gate/openstack-tox-pep8/be17b58/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/00/614400/4/gate/openstack-tox-py27/a07fa7e/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/00/614400/4/gate/openstack-tox-py36/80f0fad/ : SUCCESS in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/00/614400/4/gate/openstack-tox-py37/2c36cf7/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/00/614400/4/gate/openstack-tox-docs/261c141/html/ : SUCCESS in 2m 52s\n- build-openstack-releasenotes http://logs.openstack.org/00/614400/4/gate/build-openstack-releasenotes/93c32f3/html/ : SUCCESS in 5m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/614400/4/gate/openstack-tox-lower-constraints/c3d0ca5/ : SUCCESS in 6m 02s\n- grenade-dsvm-ceilometer http://logs.openstack.org/00/614400/4/gate/grenade-dsvm-ceilometer/aad826e/ : SUCCESS in 1h 03m 50s\n- telemetry-dsvm-integration http://logs.openstack.org/00/614400/4/gate/telemetry-dsvm-integration/cb313dd/ : SUCCESS in 54m 12s","accounts_in_message":[],"_revision_number":4},{"id":"adcf3c093be965bd90ede3c57c517d18b4ab7211","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-16 20:47:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29eee2d73812e0589e9e4b4ec9b6acd4ff3d35c1","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-17 05:46:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"79e2b4df1ba29b56238cb82b1fb392164ce03255","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-05-17 05:49:23.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nAkhil jain already cherry picked.\nhttps://review.opendev.org/#/c/659520/","accounts_in_message":[],"_revision_number":4},{"id":"44e6f4744c071f2fa2952ecdca132ec1515eda36","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-05-17 05:53:44.000000000","message":"Patch Set 4:\n\nThanks, zhurong.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"53321c1a7240967a5076f055e60092786f66b2a0","revisions":{"5873b3240c08975ae6ed21d7f7de8c620712dbb7":{"kind":"REWORK","_number":1,"created":"2018-10-31 04:20:41.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/00/614400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/00/614400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/1"}}},"commit":{"parents":[{"commit":"47e318ad6c3f16adec51f452b5b009a8e64e7788","subject":"Merge \"Don\u0027t quote {posargs} in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/47e318ad6c3f16adec51f452b5b009a8e64e7788"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2018-10-30 15:01:13.000000000","tz":330},"committer":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2018-10-31 04:20:38.000000000","tz":330},"subject":"Add framework for ceilometer-status upgrade check","message":"Add framework for ceilometer-status upgrade check\n\nThis commit adds the functionality of ceilometer-status CLI for performing\nupgrade checks as part of the Stein cycle upgrade-checkers goal.\nIt only includes a sample check which must be replaced by real checks in\nfuture.\n\nChange-Id: I6e76b74b2f02251ee39025e02de44b13568ebed2\nStory: 2003657\nTask: 27732\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/5873b3240c08975ae6ed21d7f7de8c620712dbb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/5873b3240c08975ae6ed21d7f7de8c620712dbb7"}]},"branch":"refs/heads/master"},"0485136ceada72fa652b22057cf993cfd0b95b7b":{"kind":"REWORK","_number":2,"created":"2018-10-31 04:26:51.000000000","uploader":{"_account_id":25553,"name":"Akhil jain","email":"akhiljain32@gmail.com","username":"akhil"},"ref":"refs/changes/00/614400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/00/614400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/2"}}},"commit":{"parents":[{"commit":"47e318ad6c3f16adec51f452b5b009a8e64e7788","subject":"Merge \"Don\u0027t quote {posargs} in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/47e318ad6c3f16adec51f452b5b009a8e64e7788"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2018-10-30 15:01:13.000000000","tz":330},"committer":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2018-10-31 04:26:24.000000000","tz":330},"subject":"Add framework for ceilometer-status upgrade check","message":"Add framework for ceilometer-status upgrade check\n\nThis commit adds the functionality of ceilometer-status CLI for performing\nupgrade checks as part of the Stein cycle upgrade-checkers goal.\nIt only includes a sample check which must be replaced by real checks in\nfuture.\n\nChange-Id: I6e76b74b2f02251ee39025e02de44b13568ebed2\nStory: 2003657\nTask: 27732\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/0485136ceada72fa652b22057cf993cfd0b95b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/0485136ceada72fa652b22057cf993cfd0b95b7b"}]},"branch":"refs/heads/master"},"a19b1d3044c6de0b324ca143dde65aa3a6f61987":{"kind":"REWORK","_number":3,"created":"2018-12-05 20:54:53.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/00/614400/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/00/614400/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/3"}}},"commit":{"parents":[{"commit":"47e318ad6c3f16adec51f452b5b009a8e64e7788","subject":"Merge \"Don\u0027t quote {posargs} in tox.ini\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/47e318ad6c3f16adec51f452b5b009a8e64e7788"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2018-10-30 15:01:13.000000000","tz":330},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2018-12-05 20:54:51.000000000","tz":0},"subject":"Add framework for ceilometer-status upgrade check","message":"Add framework for ceilometer-status upgrade check\n\nThis commit adds the functionality of ceilometer-status CLI for performing\nupgrade checks as part of the Stein cycle upgrade-checkers goal.\nIt only includes a sample check which must be replaced by real checks in\nfuture.\n\nChange-Id: I6e76b74b2f02251ee39025e02de44b13568ebed2\nStory: 2003657\nTask: 27732\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/a19b1d3044c6de0b324ca143dde65aa3a6f61987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/a19b1d3044c6de0b324ca143dde65aa3a6f61987"}]},"branch":"refs/heads/master"},"53321c1a7240967a5076f055e60092786f66b2a0":{"kind":"REWORK","_number":4,"created":"2019-05-16 03:15:32.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/00/614400/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/00/614400/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/00/614400/4"}}},"commit":{"parents":[{"commit":"06f66ad26282b89ffba8ed59bddc671c90297ec1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/06f66ad26282b89ffba8ed59bddc671c90297ec1"}]}],"author":{"name":"akhiljain23","email":"akhil.jain@india.nec.com","date":"2018-10-30 15:01:13.000000000","tz":330},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-05-16 03:13:43.000000000","tz":480},"subject":"Add framework for ceilometer-status upgrade check","message":"Add framework for ceilometer-status upgrade check\n\nThis commit adds the functionality of ceilometer-status CLI for performing\nupgrade checks as part of the Stein cycle upgrade-checkers goal.\nIt only includes a sample check which must be replaced by real checks in\nfuture.\n\nChange-Id: I6e76b74b2f02251ee39025e02de44b13568ebed2\nStory: 2003657\nTask: 27732\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/53321c1a7240967a5076f055e60092786f66b2a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/53321c1a7240967a5076f055e60092786f66b2a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
