)]}'
{"id":"openstack%2Fceilometer~863284","triplet_id":"openstack%2Fceilometer~stable%2Fxena~I9b34ac062d45b2b839dc9d6e0cdce696185890e4","project":"openstack/ceilometer","branch":"stable/xena","topic":"zuul_queue-stable/zed-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2022-11-07 18:09:00.000000000","reason":"Change was submitted"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2025-05-27 05:49:06.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"hashtags":[],"change_id":"I9b34ac062d45b2b839dc9d6e0cdce696185890e4","subject":"Remove lingering queue declaration","status":"MERGED","created":"2022-11-02 13:14:47.000000000","updated":"2025-05-27 05:49:06.000000000","submitted":"2022-11-07 18:09:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863284-zuul_queue-stable/zed-stable/yoga-stable/xena","cherry_pick_of_change":859795,"cherry_pick_of_patch_set":2,"meta_rev_id":"a169302ec1f72618d84787c567760599965749b6","_number":863284,"virtual_id_number":863284,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-07 18:09:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2022-11-07 14:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2022-11-07 17:15:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 14:26:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-07 08:36:12.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"state":"CC"},{"updated":"2022-11-08 13:45:53.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"state":"CC"}],"messages":[{"id":"a2a028ca38af6492a2b1db29d419e22bf33ddc1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-02 13:14:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"35f7d8247e668a1699e41539e7e616a068604002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 14:26:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f0104124ca7484385a01ac5976abfe0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/811a9b42cfc14e3fabd7f16c1dd8c64d : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff59fb6a9e9d4d5c8152f1f46ddfad2e : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67d430f07de643db820276be739aeab3 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a07e9c3d79994b5ba49158fbb982af8b : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/927c2dbd1d274c619b506f9c78ae17f2 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1d1d0c0eec4f958e30e5dcb0f5cb4e : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3be70be8585940288a583202307f1ef6 : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/412dbafa6d804d429d57f76443f9a13a : SUCCESS in 7m 05s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/0626a7b81f7d4613bf69af5827dd547c : FAILURE in 1h 05m 05s (non-voting)\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/db0f7a9b554a4ae98f6e83dd85e717cc : SUCCESS in 1h 00m 20s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/3db5eb06b41943c1a3b5db3af16fd389 : SUCCESS in 1h 03m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4cd5c7d4705e4d86aa7c4d13853e5b4f : FAILURE in 39m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8337257422ea4359b29a8a7a4cd55d48 : FAILURE in 37m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8b6d52ef983ebd60b22d135b3528014c2cbe8962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-03 05:53:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bff361c7492c757fb2d65f7e7403c3f5782c727f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 07:09:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f665652790534c74ba25a077dec9b26d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75cde4ffb2c74cc890665164bc878748 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35aa7223909c4c8aa19593bf91914bf6 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5586765bd1e45f383862072eb7541a3 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/092c56a7c92b4bffafcee6ec68fba8a0 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78a45848f7274ebd97e0a9610d1d31fe : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a447f779c7ce4a6cb842c84f8f1f38bf : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcded63e7d134b0da70e7adf3b5223ec : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9ceda3d6e9c4bc0810f67b7694d2c83 : SUCCESS in 7m 23s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/cd4f5aa582d246cd9decd66cadc246de : FAILURE in 1h 09m 00s (non-voting)\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/168f238650964eafa7aeaa008ad4bc1b : SUCCESS in 57m 01s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/68c016ca32a744aaae6e20d442f33ce9 : SUCCESS in 1h 02m 35s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/af9e8a213432441493d43a842dc4465f : SUCCESS in 49m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/942cd63fa7ed4bb3a8302bb5c85ebc3b : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e70167903695112b59ca36f2cc8f0ea10a92cf11","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-05 12:57:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7beaf62f850abc8dd16f55981c5ed28ef2f87ae6","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-07 08:36:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad18aa5f0c8e069534057d6c635083c3a79b2c33","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-07 08:44:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c1ff5423afb2b54fd86f4545ceb95a229ecf289e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-07 14:33:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c53ea251e9bdfb3b3820cf97857cd3232ca56515","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2022-11-07 17:15:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"648fe9a842f322f7b6ece54651b1848c50ba6a8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 17:16:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"06bc5bdfc3df2bd37d98050fdf016bfeb18f7258","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:09:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e28b05b660cb402899da0fabce4628a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/275c4f1675774c6ba6cf851a12b68e1c : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6d491e649f24cba878b6443fc230708 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c32d1f20d21847faa30d1c0e80281a9a : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/710a3d969ca74ce889e597642f37dab7 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c69ae0e3111948d8a6a7d129cb099dc9 : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1279fea6beec44e5bc3c09efa57176a6 : SUCCESS in 6m 14s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/f69f8d323f0743228e9a0d380b045c36 : FAILURE in 47m 47s (non-voting)\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/f4511fbd2d394a3abbe5659e39558162 : SUCCESS in 42m 45s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/d6c558ef5a8841b59084093a89bf394a : SUCCESS in 46m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ec5eeb2608e847158bbf02b0a8890f97 : SUCCESS in 32m 37s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/cf6d2e8f93014360a9b153028261e52a : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d91a82744fc893c723ceaddb05410497e37638c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:09:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5c3a0327edd144dca4733f32c2f428a621bc101a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 18:11:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/029b39e2b6f749c2a1352358ca2c6030\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d70e2973b2a4416bc2570b933516fd3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2},{"id":"18f9cbda4b04c7c7f612b685bd2359f388b349a7","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2022-11-08 13:45:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a169302ec1f72618d84787c567760599965749b6","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-05-27 05:49:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b2a9e31a2a1a29053ddaf545c986e169a0aefb7","revisions":{"02491a5ed1c5b0395ed496c6ef69777923a8ce6a":{"kind":"REWORK","_number":1,"created":"2022-11-02 13:14:47.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/84/863284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/84/863284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/84/863284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/84/863284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/84/863284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/84/863284/1"}}},"commit":{"parents":[{"commit":"274e51555f0b939efc228efe1ec34e0253742a8e","subject":"Fix broken test_discovery_with_libvirt_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/274e51555f0b939efc228efe1ec34e0253742a8e"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2022-09-29 07:18:43.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2022-11-02 13:14:47.000000000","tz":0},"subject":"Remove lingering queue declaration","message":"Remove lingering queue declaration\n\nfrom .zuul.yaml file.\n\nChange-Id: I9b34ac062d45b2b839dc9d6e0cdce696185890e4\n(cherry picked from commit 7e545b0fd9fd4a393d9e2af30f9fc370b20078c0)\n(cherry picked from commit 142cb0b6e6f6fa85634c5176c84ab15d6511fa89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/02491a5ed1c5b0395ed496c6ef69777923a8ce6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/02491a5ed1c5b0395ed496c6ef69777923a8ce6a"}]},"branch":"refs/heads/stable/xena"},"0b2a9e31a2a1a29053ddaf545c986e169a0aefb7":{"kind":"REWORK","_number":2,"created":"2022-11-03 05:53:45.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/84/863284/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/84/863284/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/84/863284/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/84/863284/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/84/863284/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/84/863284/2"}}},"commit":{"parents":[{"commit":"274e51555f0b939efc228efe1ec34e0253742a8e","subject":"Fix broken test_discovery_with_libvirt_error","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/274e51555f0b939efc228efe1ec34e0253742a8e"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2022-09-29 07:18:43.000000000","tz":120},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-03 05:53:27.000000000","tz":330},"subject":"Remove lingering queue declaration","message":"Remove lingering queue declaration\n\nfrom .zuul.yaml file.\n\nChange-Id: I9b34ac062d45b2b839dc9d6e0cdce696185890e4\n(cherry picked from commit 7e545b0fd9fd4a393d9e2af30f9fc370b20078c0)\n(cherry picked from commit 142cb0b6e6f6fa85634c5176c84ab15d6511fa89)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/0b2a9e31a2a1a29053ddaf545c986e169a0aefb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/0b2a9e31a2a1a29053ddaf545c986e169a0aefb7"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[]}
