)]}'
{"id":"openstack%2Fceilometer~96944","triplet_id":"openstack%2Fceilometer~stable%2Ficehouse~Ia1bfa1bd24989681db1d2f385defc12e69a01f8d","project":"openstack/ceilometer","branch":"stable/icehouse","topic":"(detached","hashtags":[],"change_id":"Ia1bfa1bd24989681db1d2f385defc12e69a01f8d","subject":"remove token from notifier middleware","status":"MERGED","created":"2014-05-30 21:11:32.000000000","updated":"2014-06-26 13:00:48.000000000","submitted":"2014-06-26 13:00:48.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5ef27a927ac591f92ac9070c1e6289f2434fcbd4","_number":96944,"virtual_id_number":96944,"owner":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2014-06-26 13:00:48.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":6676,"name":"zqfan","email":"aji.zqfan@gmail.com","username":"aji-zqfan"},{"value":0,"_account_id":3012,"name":"Dina Belova","email":"dbelova@mirantis.com","username":"dbelova"},{"value":0,"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"value":0,"date":"2014-06-25 21:21:46.000000000","_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"value":0,"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},{"value":0,"_account_id":2284,"name":"Eoghan Glynn","email":"eglynn@redhat.com","username":"eglynn"},{"value":0,"_account_id":1215,"name":"Adam Gandelman","inactive":true}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"recommended":{"_account_id":3012,"name":"Dina Belova","email":"dbelova@mirantis.com","username":"dbelova"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":1,"date":"2014-05-31 04:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6676,"name":"zqfan","email":"aji.zqfan@gmail.com","username":"aji-zqfan"},{"value":1,"date":"2014-06-26 10:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":3012,"name":"Dina Belova","email":"dbelova@mirantis.com","username":"dbelova"},{"value":2,"date":"2014-06-26 11:33:14.000000000","_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"value":0,"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"value":1,"date":"2014-06-10 16:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},{"value":2,"date":"2014-06-25 19:31:20.000000000","_account_id":2284,"name":"Eoghan Glynn","email":"eglynn@redhat.com","username":"eglynn"},{"value":0,"_account_id":1215,"name":"Adam Gandelman","inactive":true}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2014-06-26 11:31:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"_account_id":6676,"name":"zqfan","email":"aji.zqfan@gmail.com","username":"aji-zqfan"},{"value":0,"_account_id":3012,"name":"Dina Belova","email":"dbelova@mirantis.com","username":"dbelova"},{"value":1,"date":"2014-06-26 11:35:39.000000000","_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"value":0,"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"value":0,"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},{"value":1,"date":"2014-06-26 11:34:40.000000000","_account_id":2284,"name":"Eoghan Glynn","email":"eglynn@redhat.com","username":"eglynn"},{"value":0,"date":"2014-06-16 13:46:25.000000000","_account_id":1215,"name":"Adam Gandelman","inactive":true}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1215,"name":"Adam Gandelman","inactive":true},{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":2284,"name":"Eoghan Glynn","email":"eglynn@redhat.com","username":"eglynn"},{"_account_id":3012,"name":"Dina Belova","email":"dbelova@mirantis.com","username":"dbelova"},{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"_account_id":6676,"name":"zqfan","email":"aji.zqfan@gmail.com","username":"aji-zqfan"},{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-05-31 04:01:55.000000000","updated_by":{"_account_id":6676,"name":"zqfan","email":"aji.zqfan@gmail.com","username":"aji-zqfan"},"reviewer":{"_account_id":6676,"name":"zqfan","email":"aji.zqfan@gmail.com","username":"aji-zqfan"},"state":"REVIEWER"},{"updated":"2014-06-10 16:40:56.000000000","updated_by":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"reviewer":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"state":"REVIEWER"},{"updated":"2014-06-16 13:46:25.000000000","updated_by":{"_account_id":1215,"name":"Adam Gandelman","inactive":true},"reviewer":{"_account_id":1215,"name":"Adam Gandelman","inactive":true},"state":"REVIEWER"},{"updated":"2014-06-26 10:00:42.000000000","updated_by":{"_account_id":3012,"name":"Dina Belova","email":"dbelova@mirantis.com","username":"dbelova"},"reviewer":{"_account_id":3012,"name":"Dina Belova","email":"dbelova@mirantis.com","username":"dbelova"},"state":"REVIEWER"},{"updated":"2014-06-26 11:31:21.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2014-06-26 11:34:40.000000000","updated_by":{"_account_id":2284,"name":"Eoghan Glynn","email":"eglynn@redhat.com","username":"eglynn"},"reviewer":{"_account_id":2284,"name":"Eoghan Glynn","email":"eglynn@redhat.com","username":"eglynn"},"state":"REVIEWER"},{"updated":"2014-06-26 11:35:39.000000000","updated_by":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"reviewer":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"state":"REVIEWER"},{"updated":"2014-06-26 13:00:48.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"843b6ace3635cff9473d6e383ac0c8370c4d7ebf","author":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"date":"2014-05-30 21:11:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ceb8a501d20b1632cfc2fff85632a9582ef169d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-30 21:11:33.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"9ee675f6fafc5b16785487e05f2b60625dfeace9","author":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"date":"2014-05-30 21:20:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea91d853f3ec159fe6637c68f8ce8f3a26820fb4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-30 21:20:56.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"86cd6bf1b79120461759fc96182c4267b9591798","author":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"date":"2014-05-30 21:21:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b37496192f2d9ca012a7c2e7cb8849bc63402138","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-30 21:21:54.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"5f0818784fec3260bf6fa9003dbbee8e2dafdc19","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-30 22:17:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-ceilometer-pep8 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-pep8/0d4d89b : SUCCESS in 4m 16s\n- gate-ceilometer-docs http://docs-draft.openstack.org/44/96944/3/check/gate-ceilometer-docs/355dd4e/doc/build/html/ : SUCCESS in 4m 18s\n- gate-ceilometer-python26 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-python26/b478bcb : SUCCESS in 8m 45s\n- gate-ceilometer-python27 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-python27/69f0839 : SUCCESS in 5m 30s\n- check-devstack-dsvm-cells http://logs.openstack.org/44/96944/3/check/check-devstack-dsvm-cells/3acdcb9 : SUCCESS in 28m 35s\n- check-tempest-dsvm-full http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-full/83e8e54 : SUCCESS in 55m 45s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-postgres-full/0d24c6d : SUCCESS in 44m 56s\n- check-tempest-dsvm-neutron http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-neutron/1af3471 : SUCCESS in 55m 04s\n- check-swift-dsvm-functional http://logs.openstack.org/44/96944/3/check/check-swift-dsvm-functional/6e89a40 : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":3},{"id":"bcb4d3f82112b4e73828cabef13900e138101adc","author":{"_account_id":6676,"name":"zqfan","email":"aji.zqfan@gmail.com","username":"aji-zqfan"},"date":"2014-05-31 04:01:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"49433304a257a1419d705ad5509e3544b8ba4432","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2014-06-10 16:40:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3e5cdfaa45d14fc395a120b9b5f55a9e522dd342","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-10 16:40:57.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"3fbc4442db7d9989d0e16ce35f51aaa56ee68153","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-10 18:33:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-ceilometer-pep8 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-pep8/75c5a02 : SUCCESS in 3m 46s\n- gate-ceilometer-docs http://docs-draft.openstack.org/44/96944/3/check/gate-ceilometer-docs/e9556cc/doc/build/html/ : SUCCESS in 3m 36s\n- gate-ceilometer-python26 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-python26/50d6646 : SUCCESS in 7m 05s\n- gate-ceilometer-python27 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-python27/5d5e14f : SUCCESS in 5m 32s\n- check-devstack-dsvm-cells http://logs.openstack.org/44/96944/3/check/check-devstack-dsvm-cells/019a569 : SUCCESS in 12m 37s\n- check-tempest-dsvm-full http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-full/81aab5c : SUCCESS in 54m 39s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-postgres-full/f129c44 : SUCCESS in 47m 00s\n- check-tempest-dsvm-neutron http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-neutron/713de84 : SUCCESS in 58m 19s\n- check-swift-dsvm-functional http://logs.openstack.org/44/96944/3/check/check-swift-dsvm-functional/0b0f571 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":3},{"id":"7bc00cd02018de6717c8f59d5910e7292e5a1780","author":{"_account_id":2284,"name":"Eoghan Glynn","email":"eglynn@redhat.com","username":"eglynn"},"date":"2014-06-25 19:31:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"af7540e939ad848dc19af0fd764cb4d66dfa37f0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-25 19:31:21.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"b7e79c909ba77388b81d203b5b3bd9d6f4ed5ce6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-25 20:32:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-ceilometer-pep8 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-pep8/809e0b8 : SUCCESS in 5m 00s\n- gate-ceilometer-docs http://docs-draft.openstack.org/44/96944/3/check/gate-ceilometer-docs/db5149b/doc/build/html/ : SUCCESS in 6m 19s\n- gate-ceilometer-python26 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-python26/8e219f5 : FAILURE in 2m 57s\n- gate-ceilometer-python27 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-python27/5ba77ad : SUCCESS in 6m 29s\n- check-tempest-dsvm-full http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-full/8c6c503 : SUCCESS in 45m 26s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-postgres-full/d3ebec8 : SUCCESS in 45m 07s\n- check-tempest-dsvm-neutron http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-neutron/499d2a6 : SUCCESS in 49m 46s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-neutron-heat-slow/54eb4f5 : SUCCESS in 26m 51s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/44/96944/3/check/check-grenade-dsvm/3858630 : SUCCESS in 37m 59s\n- check-devstack-dsvm-cells http://logs.openstack.org/44/96944/3/check/check-devstack-dsvm-cells/b54fec7 : SUCCESS in 13m 29s\n- check-swift-dsvm-functional http://logs.openstack.org/44/96944/3/check/check-swift-dsvm-functional/332b50f : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":3},{"id":"2fa7c818ee7111521d6df6d12595a85ebce584fd","author":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"date":"2014-06-25 21:21:46.000000000","message":"Patch Set 3:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":3},{"id":"eee71629844c8743643b13b87785a0a2385cc3e4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-25 21:21:47.000000000","message":"Patch Set 3: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"219b85e94f365b300c586aa0c62a41450fbbf269","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-25 22:28:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded.\n\n- gate-ceilometer-pep8 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-pep8/4d11b2f : SUCCESS in 4m 27s\n- gate-ceilometer-docs http://docs-draft.openstack.org/44/96944/3/check/gate-ceilometer-docs/3c92b53/doc/build/html/ : SUCCESS in 5m 29s\n- gate-ceilometer-python26 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-python26/aace135 : SUCCESS in 6m 30s\n- gate-ceilometer-python27 http://logs.openstack.org/44/96944/3/check/gate-ceilometer-python27/4a49ec0 : SUCCESS in 4m 32s\n- check-tempest-dsvm-full http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-full/f8a5e2a : SUCCESS in 42m 36s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-postgres-full/2096d97 : SUCCESS in 41m 53s\n- check-tempest-dsvm-neutron http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-neutron/31c39e1 : SUCCESS in 54m 47s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/44/96944/3/check/check-tempest-dsvm-neutron-heat-slow/95b06cc : SUCCESS in 27m 53s\n- check-grenade-dsvm http://logs.openstack.org/44/96944/3/check/check-grenade-dsvm/1e49a07 : SUCCESS in 37m 57s\n- check-devstack-dsvm-cells http://logs.openstack.org/44/96944/3/check/check-devstack-dsvm-cells/15c8075 : SUCCESS in 15m 24s\n- check-swift-dsvm-functional http://logs.openstack.org/44/96944/3/check/check-swift-dsvm-functional/9838235 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":3},{"id":"7a814786ffbdfd468558af2dbdc345c08e1e91e2","author":{"_account_id":3012,"name":"Dina Belova","email":"dbelova@mirantis.com","username":"dbelova"},"date":"2014-06-26 10:00:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1e47519218a57e4a00f1e886bc26eaa634877069","author":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"date":"2014-06-26 11:33:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1d12b084ef21b259cf753f20618473131d277122","author":{"_account_id":2284,"name":"Eoghan Glynn","email":"eglynn@redhat.com","username":"eglynn"},"date":"2014-06-26 11:34:40.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"76103cca24e376a4c4636659d1add57493775f5c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-26 11:34:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"f55ded7e77a9c0b9ad37c4e10d6d35ea8b644c7f","author":{"_account_id":1812,"name":"p-draigbrady","email":"P@draigBrady.com","username":"p-draigbrady"},"date":"2014-06-26 11:35:39.000000000","message":"Patch Set 3: Workflow+1\n\nOpenStack Security Advisory:2014-021\nCVE: CVE-2014-4615\nappropriate for stable\nthanks","accounts_in_message":[],"_revision_number":3},{"id":"3ddafce64b2786455751c3bc8bc6a0bc49ac1bb3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-26 13:00:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded.\n\n- gate-ceilometer-docs http://docs-draft.openstack.org/44/96944/3/gate/gate-ceilometer-docs/7e40c29/doc/build/html/ : SUCCESS in 5m 27s\n- gate-ceilometer-pep8 http://logs.openstack.org/44/96944/3/gate/gate-ceilometer-pep8/96f453a : SUCCESS in 3m 00s\n- gate-ceilometer-python26 http://logs.openstack.org/44/96944/3/gate/gate-ceilometer-python26/fb61bb9 : SUCCESS in 5m 00s\n- gate-ceilometer-python27 http://logs.openstack.org/44/96944/3/gate/gate-ceilometer-python27/7ea9c57 : SUCCESS in 6m 16s\n- gate-tempest-dsvm-full http://logs.openstack.org/44/96944/3/gate/gate-tempest-dsvm-full/2b13c7b : SUCCESS in 45m 17s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/44/96944/3/gate/gate-tempest-dsvm-postgres-full/e61e497 : SUCCESS in 41m 14s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/44/96944/3/gate/gate-tempest-dsvm-neutron/b1d0e8e : SUCCESS in 52m 40s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/44/96944/3/gate/gate-tempest-dsvm-neutron-heat-slow/a243c43 : SUCCESS in 23m 46s\n- gate-grenade-dsvm http://logs.openstack.org/44/96944/3/gate/gate-grenade-dsvm/9d71c85 : SUCCESS in 37m 29s\n- gate-devstack-dsvm-cells http://logs.openstack.org/44/96944/3/gate/gate-devstack-dsvm-cells/1072c96 : SUCCESS in 15m 08s\n- gate-swift-dsvm-functional http://logs.openstack.org/44/96944/3/gate/gate-swift-dsvm-functional/8e957d6 : SUCCESS in 14m 39s","accounts_in_message":[],"_revision_number":3},{"id":"80d3c44dc031d3c363ff28df57756bfccd9c3da7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-26 13:00:48.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b6454f9f4e0585949ab68a91ed405755438d76e","revisions":{"8fa8b0cb116baeae0931635204e44e6d39a07334":{"kind":"REWORK","_number":1,"created":"2014-05-30 21:11:32.000000000","uploader":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"ref":"refs/changes/44/96944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/44/96944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/1"}}},"commit":{"parents":[{"commit":"94c12fd76c8b92a4d25f952fc760466c22213fe2","subject":"Adapt failing tests for latest wsme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/94c12fd76c8b92a4d25f952fc760466c22213fe2"}]}],"author":{"name":"gordon chung","email":"chungg@ca.ibm.com","date":"2014-05-30 21:11:18.000000000","tz":-240},"committer":{"name":"gordon chung","email":"chungg@ca.ibm.com","date":"2014-05-30 21:11:18.000000000","tz":-240},"subject":"remove token from notifier middleware","message":"remove token from notifier middleware\n\nnotifier middleware is capturing token and sending it to MQ. this\nis not advisable so we should filter it out.\n\nChange-Id: Ia1bfa1bd24989681db1d2f385defc12e69a01f8d\nCloses-Bug: #1321080\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/8fa8b0cb116baeae0931635204e44e6d39a07334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/8fa8b0cb116baeae0931635204e44e6d39a07334"}]},"branch":"refs/heads/stable/icehouse"},"2f6c5ed8d8ab9ef9a25b6aed0b65f8f0f9e7f0dc":{"kind":"REWORK","_number":2,"created":"2014-05-30 21:20:54.000000000","uploader":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"ref":"refs/changes/44/96944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/44/96944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/2"}}},"commit":{"parents":[{"commit":"94c12fd76c8b92a4d25f952fc760466c22213fe2","subject":"Adapt failing tests for latest wsme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/94c12fd76c8b92a4d25f952fc760466c22213fe2"}]}],"author":{"name":"gordon chung","email":"chungg@ca.ibm.com","date":"2014-05-30 21:11:18.000000000","tz":-240},"committer":{"name":"gordon chung","email":"chungg@ca.ibm.com","date":"2014-05-30 21:20:46.000000000","tz":-240},"subject":"remove token from notifier middleware","message":"remove token from notifier middleware\n\nnotifier middleware is capturing token and sending it to MQ. this\nis not advisable so we should filter it out.\n\nChange-Id: Ia1bfa1bd24989681db1d2f385defc12e69a01f8d\nCloses-Bug: #1321080\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/2f6c5ed8d8ab9ef9a25b6aed0b65f8f0f9e7f0dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/2f6c5ed8d8ab9ef9a25b6aed0b65f8f0f9e7f0dc"}]},"branch":"refs/heads/stable/icehouse"},"2b6454f9f4e0585949ab68a91ed405755438d76e":{"kind":"REWORK","_number":3,"created":"2014-05-30 21:21:52.000000000","uploader":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"ref":"refs/changes/44/96944/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/44/96944/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/44/96944/3"}}},"commit":{"parents":[{"commit":"94c12fd76c8b92a4d25f952fc760466c22213fe2","subject":"Adapt failing tests for latest wsme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/94c12fd76c8b92a4d25f952fc760466c22213fe2"}]}],"author":{"name":"gordon chung","email":"chungg@ca.ibm.com","date":"2014-05-30 21:11:18.000000000","tz":-240},"committer":{"name":"gordon chung","email":"chungg@ca.ibm.com","date":"2014-05-30 21:21:47.000000000","tz":-240},"subject":"remove token from notifier middleware","message":"remove token from notifier middleware\n\nnotifier middleware is capturing token and sending it to MQ. this\nis not advisable so we should filter it out.\n\nChange-Id: Ia1bfa1bd24989681db1d2f385defc12e69a01f8d\nCloses-Bug: #1321080\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/2b6454f9f4e0585949ab68a91ed405755438d76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/2b6454f9f4e0585949ab68a91ed405755438d76e"}]},"branch":"refs/heads/stable/icehouse"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
