)]}'
{"id":"openstack%2Fceilometer~995389","triplet_id":"openstack%2Fceilometer~master~I86f657402dce2830cf1c2e438f7b4207b166bcf4","project":"openstack/ceilometer","branch":"master","attention_set":{"28356":{"account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"last_update":"2026-08-07 07:11:51.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"removed_from_attention_set":{"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-09-10 07:45:57.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2026-08-07 07:11:51.000000000","reason":"\u003cGERRIT_ACCOUNT_4264\u003e replied on the change","reason_account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}},"hashtags":[],"change_id":"I86f657402dce2830cf1c2e438f7b4207b166bcf4","subject":"Make each dynamic pollster definition have its own discovery instance","status":"NEW","created":"2026-06-29 17:45:37.000000000","updated":"2026-09-10 07:45:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":17,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"8b680dec1aad9c200466e3ef97f07c234d111b27","_number":995389,"virtual_id_number":995389,"owner":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-06 23:12:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2026-08-07 07:11:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-29 18:56:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-09 13:00:43.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"CC"},{"updated":"2026-07-09 13:26:23.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2026-08-06 12:12:15.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-09-08 09:08:08.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"CC"}],"messages":[{"id":"4f4a4905dbfe010a7480b50fb0180fad74a52b6d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-06-29 17:45:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82cfea465ac17a6103c3497993fbefefb090382e","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-06-29 17:46:39.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"efb6c1371cba3c3d5a280ec4d635b2da5278def5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 18:56:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f63a40ca9f314a9bb57a37d9c7b9cff3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f51d399254446d0a3aa5bedbab321f4 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98731aa2316c406a8b66827b428988d3 : FAILURE in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28391aec715449299e08c9035ee656f9 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f2bff996c064b5e89a5fbd5461bb02e : SUCCESS in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7739b30eefd340e0a137dc89c1d1b8b1 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868a494c27e247d0a2bce2d6114f789f : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/317c8047d26249ba87d84feb1823cdd5 : SUCCESS in 4m 19s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/4ada62ff630d4e2aa519812f0dd1c447 : SUCCESS in 47m 48s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/c3fe10ac7f414fa6ade9d5a3571a5941 : SUCCESS in 1h 02m 23s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/197bc18391c4493ebde1c5ffa898c298 : SUCCESS in 1h 04m 57s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9c77c84fc7d14dcdb1378d09003f98ea : SUCCESS in 53m 40s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/a7d1b6b339a5405fbc1d418b7d296c90 : SUCCESS in 43m 06s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f5a0c3e1d5dc467bb703e6cce2fdaf0e : SUCCESS in 32m 51s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/ff6665382b8449fd8c866bb56f165628 : SUCCESS in 27m 05s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"9bffe68173c0a5877b290054b43eba5756b2659d","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-02 01:56:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"348a55c10e2cfb736b0c5735732eb13218e059db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 03:14:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/926ad9f0ebbf4a339692bbfe23c4a27d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7b81f5d3a9c47f8b6571519f7b86966 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/361e28d128c04bbf96246837897768fd : FAILURE in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f37ce2a0cb34e008ba53fcc74c0c741 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6b481799b0241208e43d388c7cc35d2 : SUCCESS in 5m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/418498669c4d4545bdf2e6ab1532ca92 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28eb8f47f91045da922adc4f5af60c67 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5125e2685714e6ba96c2ef787a99078 : SUCCESS in 7m 19s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/61efde253efe41a180697162c7b50e91 : SUCCESS in 1h 12m 31s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/fb8f2656c71b4457b9f48cc9ef0b0dd0 : SUCCESS in 1h 03m 52s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/613aeb89a91549bf80eb24503a8b4399 : SUCCESS in 1h 02m 20s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/590a2ba62f12462fab5079c7ee9785e0 : SUCCESS in 44m 34s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/49d4c2aa78d5425f9af1eeb115aeee84 : SUCCESS in 46m 26s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e70e4e418aa4437f802a0b5bacbd7df8 : SUCCESS in 52m 06s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/62315538ab154d59ade6f03d8d9867f3 : SUCCESS in 50m 23s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"862c6ccc633ceea37faf0374d1f807744b6b0a3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-02 11:46:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"b188104ecacbe86ad44bff479039be006d56808d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 13:08:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56c36b3f243f4255a8f2d468c67d1e2c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b67aab30b184970bf11d6bafcabeb6f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19e854c219164081addc0edbb1c3fc51 : FAILURE in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/349126a48f364fa5b39f4accdbb75a19 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fb8900e278134f63872777932a9802c4 : SUCCESS in 7m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/57207783ac164efd944a7689bcd6a8c0 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/429627050a6c49af8511f4f3d1da69a4 : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/680aadea000240f9b0f14f3891efdff5 : SUCCESS in 6m 05s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/89003ff6d84a424c8293302defbbb4a4 : SUCCESS in 1h 14m 18s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/02a38852aad54106b597be1e58b02ac1 : SUCCESS in 39m 26s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/2f256783f6c842a7b02f44a1801edfdc : SUCCESS in 40m 51s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7e4baa9371684ca0bb34f5920b893be1 : SUCCESS in 45m 39s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/8026613a6bda4b18989aa8d1805d936c : SUCCESS in 26m 22s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a60f3e1552a444f0b3f1ed507fb59732 : SUCCESS in 50m 47s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/2522d911178b4f4a91f95b1aa9ded718 : SUCCESS in 42m 15s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"457a81e74fbd4c48771206f9f6ecdd05c0e1c8a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-02 18:19:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c5228e64c2ca74b6da11b1d23d00202feb1f8306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 19:32:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51c781ccf754e94b9caaf40162ca6ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/721b6bfef3c640f7be7ccf6960e33831 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af864a51b293446ba8aa182e02f45ee2 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0dc4564d4d0a43b5823a3c4a55bbf42a : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e66da16d7adf42cc9edecb7b5b7cea78 : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/18acdf86040045e7a4035ca3f9f7e461 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11dc0d0c12cf49b39e5d67dedab2a181 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/971109294ded4fa19b9ac53909d32b98 : SUCCESS in 7m 08s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/8a7a0901dfd14340a96f410972c717d9 : SUCCESS in 1h 03m 55s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/e93e9172e317456ab98d7a9505a168f6 : SUCCESS in 36m 48s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/191ed01ca5d14a72b913da101bfed4cb : SUCCESS in 1h 05m 47s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/21589ecff5be457eb69d071eb9932930 : SUCCESS in 1h 02m 54s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/497f4f6382664950aee31351420bbb35 : SUCCESS in 45m 35s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0e54d4a65b86470eba342e5426562609 : SUCCESS in 34m 37s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/9716c254e7564313b6726e63c9465d7d : SUCCESS in 23m 10s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"96f51ccc51b247ecc39b0b09385a403701221ae7","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-07-09 13:00:43.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"977a1c57f64d86b0b03e7cbedffc03fbce788fbf","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-07-09 13:26:23.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"4eae274c998d49a64a1721cdd743c968014e3d13","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-07-09 13:45:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e0308e03c0ee855d540f4e1b76d92eb8b53b802d","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-13 19:32:31.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"24383366a6b036a4aa946b2483b2795015e0afeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-13 19:32:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c6bd4925b21e23b138f34c1ce3d36b4848894f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-13 19:44:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e5fbaf4832af49dca60da671085fa7ca35a334a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 21:05:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/391764c612a14c8ebfbd67d5e7d6e31f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a94bf71e10f4915a151b980d89d454c : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcea19994d5b48b9ab20e08f817c2fb3 : FAILURE in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/034447a671ae4b1b89af1ac1697fb8a2 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14367bf64a8b4fb29154fd56cea633a6 : FAILURE in 4m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/77cbb79dd6ad475d8ac7170ab05b715e : FAILURE in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371bdee9a0a94df799579b514627b8df : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4900f30b23f4643899688d69a188d40 : SUCCESS in 6m 08s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/6b3d4627859047a49d94fa8400b850c0 : SUCCESS in 38m 50s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/8ae1be2376504229bb936f6bcbf82bbe : SUCCESS in 1h 02m 15s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/2574349b22c74f919eea864a1b5f1feb : SUCCESS in 1h 14m 22s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4ef4f903e8f94ea09362b3586de27fa7 : SUCCESS in 55m 37s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/2d8177ce0e3d44bba0b55558e181953d : SUCCESS in 37m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f23719a588c245f0afaa79ed10597323 : SUCCESS in 50m 00s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/665f3833c93c4afba766b1ea1296c8b3 : SUCCESS in 32m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"c665bd58c7f461a09febc65968d21d0cd0433bfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-14 11:58:51.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"b26996272f7e9b81a0de2456c324fd999202bb5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 13:15:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddf6fb73ebb3436a8f19be50b4933567\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09340b302dfe40839cd60b947cbb9e78 : FAILURE in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad038fa456545afadde87453f9a2ee5 : FAILURE in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e9a21c556b749bab4f7d77762a5040f : SUCCESS in 7m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe1660120f6343599bfc827ba527181c : FAILURE in 5m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f90213e6d30441aa182615957252c6e : FAILURE in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2baa0001772f417ab5f65ace5a533947 : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7faded116588414ba69827687f46aab4 : SUCCESS in 5m 17s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/e94a3592f4f344558256ba2cfc652284 : SUCCESS in 1h 07m 08s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/686ce47477134ff6af1ac9ff8f8fc16b : SUCCESS in 1h 10m 44s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/a3cf79e85be94f1898eb7e8c159028f5 : SUCCESS in 1h 10m 38s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/384f0f20557648938c1c9ab9175b66f2 : SUCCESS in 52m 27s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/07f9befb07ef4b29b71a42bd1d3ee703 : SUCCESS in 42m 19s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6fb76c7a902544208de1d2557009e3e4 : SUCCESS in 45m 24s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/c5888b8c37f346fab1edc584e1f50ae7 : SUCCESS in 44m 46s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"399826f9d05ab39faadc327a7d10bc9858d32408","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-14 13:51:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"f4104f3069b714f5f8e3108e87ae6148b9a85ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 15:07:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8afcbd5b2634556ab31ec4d08141aef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a691649e2169482481d8e38895878bf3 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d6e2db589846acbb8d8eb8e4840d96 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc1d0c35827949e2910429fae9fbfc85 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/79c02795244946e7b3346df624a56569 : SUCCESS in 6m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97de7e5ea3ba4d13a4240a1ccdbbc227 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b38e0a94df7449859b2033dce11ae16d : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ba2af00ea2c4d2c94dbcdc01b6bda7e : SUCCESS in 7m 06s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/988bbd03636f434d9135ffa5d2e5eb9d : SUCCESS in 1h 11m 20s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/05822546efcd4c0dabcd20ca90773ab2 : SUCCESS in 42m 01s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/b8faef6a054e4faeb5f1ab2ce68ce495 : SUCCESS in 47m 59s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/279651acab8d4993bb650bf53a0a9390 : SUCCESS in 56m 57s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/abbd6484838044a7a69a010fe46b0400 : SUCCESS in 37m 33s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/b0dd1796f67c454eaf9ec4b8e799fd35 : SUCCESS in 53m 42s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/6ec533866c494b56960663a2c8bce720 : SUCCESS in 21m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"df65e52dae77401df2a8136cacebb2e9f0aaa68e","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-07-14 17:15:40.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"00c08cb13161cfea6015be11ec229a7a5089685c","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-07-15 07:31:20.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"96d991da4f80a814faf2e2e02e8da326dc784123","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-08-06 12:12:15.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a5dc6c3baaac17a72dc977425e7a0a86c68b883c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-08-06 22:03:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0d573743659a0f8143610fb158fc3d9b0040ecf5","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-08-06 22:03:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5ffdf03290fe0cf10d2e1f7fb1ef71e850a0928f","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2026-08-06 22:04:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b2208d660cda19dc934ddc6f5c42c030e61876a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 23:12:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a1d0daf5fa645c7949002033567c333\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73cecc88942f4faa919b90d040c53a7b : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d9cd6349f64111bb2422c35948556a : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8286d33d818b4a7290ef01facc3d13ab : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea3c1ba6451e4e898deb903968cd283e : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/88f1dd5fe3f442bd884988556941c685 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9432a760b9434ac8b0ab3fef72d1cd07 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cb15f5e90f34ddbb7e5bcc60f8f2279 : SUCCESS in 5m 32s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/266dade52a384ebb9a82d6f14fdbfca6 : SUCCESS in 48m 21s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/d89c332628fe41a780ec9498b29f4fe7 : SUCCESS in 50m 33s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/0b413bff6941470ab4918fb1b8672b4c : SUCCESS in 1h 05m 56s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/69bb885161554d29857436ff94aaa6f7 : SUCCESS in 29m 38s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/609d06f9aed141e287276e6f65319df9 : SUCCESS in 26m 54s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/457bdf77334049558812230000ac296d : SUCCESS in 36m 40s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/7cd02f93e3484c3183847f27f0dbb975 : SUCCESS in 44m 07s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8},{"id":"44a18d25d1838d1019a3062a7297b7feea9a92b3","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-08-07 07:11:51.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"28fcbb0511ca043a925223ef7b8320bd37a41506","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-09-08 09:08:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8b680dec1aad9c200466e3ef97f07c234d111b27","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-09-10 07:45:57.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"3b83d59fefbaf9075b579f0c07ec705c9a467292","revisions":{"c70d873004f57feb8a5334de1419fcf14bfe0969":{"kind":"REWORK","_number":1,"created":"2026-06-29 17:45:37.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/89/995389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/89/995389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/1"}}},"commit":{"parents":[{"commit":"20b6241fef196fea01ee81b6e8ab3d9534d76747","subject":"Merge \"ruff: Enable LOG checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/20b6241fef196fea01ee81b6e8ab3d9534d76747"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:44:36.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:45:17.000000000","tz":-180},"subject":"Make each dynamic pollster definition have its own discovery instance","message":"Make each dynamic pollster definition have its own discovery instance\n\nWhen coordination is enabled in Ceilometer, only one control node will\nexecute the pollster (either the hardcoded one or the dynamic pollster).\nThe synchronization is performed using the \"pooling.yml\" configuration\nin the \"ceilometer.polling.manager.PollingTask._internal_pollster_run\"\nmethod using the pollster name.\n\nHowever, the check (and further pollster execution) is only executed if\nthe code finds \"polling_resources\", which are the return of the\ndiscovery process. The discovery process also has a lock/coordination\nprocess inside it, and it uses the group of the discovery, which in the\ncase of the endpoint and Barbican discovery process, is a hardcoded\nstring called \"global\". This, in turn, locks all the execution of the\ndiscovery mechanism to the very first control node that executes it.\nAlso, it makes all pollsters that depend/need these discoveries be\nexecuted in the same control node.\n\nThis patch proposes a new structure, so each dynamic pollster has its\nown discovery instance. This, in turn, enables us to better distribute\nthe execution of the pollster in the cloud, as they are dependent on the\ndiscovery, which has a lock/coordination process as well.\n\nChange-Id: I86f657402dce2830cf1c2e438f7b4207b166bcf4\nSigned-off-by: Rafael Weingärtner \u003crafael@apache.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/c70d873004f57feb8a5334de1419fcf14bfe0969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/c70d873004f57feb8a5334de1419fcf14bfe0969"}]},"branch":"refs/heads/master"},"d465eeb7671c8a516f45cb74c374b4fb4bef1e01":{"kind":"REWORK","_number":2,"created":"2026-07-02 11:46:44.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/89/995389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/89/995389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/2"}}},"commit":{"parents":[{"commit":"20b6241fef196fea01ee81b6e8ab3d9534d76747","subject":"Merge \"ruff: Enable LOG checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/20b6241fef196fea01ee81b6e8ab3d9534d76747"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:44:36.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-07-02 11:46:13.000000000","tz":-180},"subject":"Make each dynamic pollster definition have its own discovery instance","message":"Make each dynamic pollster definition have its own discovery instance\n\nWhen coordination is enabled in Ceilometer, only one control node will\nexecute the pollster (either the hardcoded one or the dynamic pollster).\nThe synchronization is performed using the \"pooling.yml\" configuration\nin the \"ceilometer.polling.manager.PollingTask._internal_pollster_run\"\nmethod using the pollster name.\n\nHowever, the check (and further pollster execution) is only executed if\nthe code finds \"polling_resources\", which are the return of the\ndiscovery process. The discovery process also has a lock/coordination\nprocess inside it, and it uses the group of the discovery, which in the\ncase of the endpoint and Barbican discovery process, is a hardcoded\nstring called \"global\". This, in turn, locks all the execution of the\ndiscovery mechanism to the very first control node that executes it.\nAlso, it makes all pollsters that depend/need these discoveries be\nexecuted in the same control node.\n\nThis patch proposes a new structure, so each dynamic pollster has its\nown discovery instance. This, in turn, enables us to better distribute\nthe execution of the pollster in the cloud, as they are dependent on the\ndiscovery, which has a lock/coordination process as well.\n\nChange-Id: I86f657402dce2830cf1c2e438f7b4207b166bcf4\nSigned-off-by: Rafael Weingärtner \u003crafael@apache.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/d465eeb7671c8a516f45cb74c374b4fb4bef1e01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/d465eeb7671c8a516f45cb74c374b4fb4bef1e01"}]},"branch":"refs/heads/master"},"89a38ad4c58ec351fee503a96590c864277a887f":{"kind":"REWORK","_number":3,"created":"2026-07-02 18:19:28.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/89/995389/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/89/995389/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/3"}}},"commit":{"parents":[{"commit":"20b6241fef196fea01ee81b6e8ab3d9534d76747","subject":"Merge \"ruff: Enable LOG checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/20b6241fef196fea01ee81b6e8ab3d9534d76747"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:44:36.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-07-02 18:18:58.000000000","tz":-180},"subject":"Make each dynamic pollster definition have its own discovery instance","message":"Make each dynamic pollster definition have its own discovery instance\n\nWhen coordination is enabled in Ceilometer, only one control node will\nexecute the pollster (either the hardcoded one or the dynamic pollster).\nThe synchronization is performed using the \"pooling.yml\" configuration\nin the \"ceilometer.polling.manager.PollingTask._internal_pollster_run\"\nmethod using the pollster name.\n\nHowever, the check (and further pollster execution) is only executed if\nthe code finds \"polling_resources\", which are the return of the\ndiscovery process. The discovery process also has a lock/coordination\nprocess inside it, and it uses the group of the discovery, which in the\ncase of the endpoint and Barbican discovery process, is a hardcoded\nstring called \"global\". This, in turn, locks all the execution of the\ndiscovery mechanism to the very first control node that executes it.\nAlso, it makes all pollsters that depend/need these discoveries be\nexecuted in the same control node.\n\nThis patch proposes a new structure, so each dynamic pollster has its\nown discovery instance. This, in turn, enables us to better distribute\nthe execution of the pollster in the cloud, as they are dependent on the\ndiscovery, which has a lock/coordination process as well.\n\nChange-Id: I86f657402dce2830cf1c2e438f7b4207b166bcf4\nSigned-off-by: Rafael Weingärtner \u003crafael@apache.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/89a38ad4c58ec351fee503a96590c864277a887f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/89a38ad4c58ec351fee503a96590c864277a887f"}]},"branch":"refs/heads/master"},"fbac3d9266bedbb1cf1a58be916e9a9fcb3cc69b":{"kind":"REWORK","_number":4,"created":"2026-07-13 19:32:33.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/89/995389/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/89/995389/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/4"}}},"commit":{"parents":[{"commit":"20b6241fef196fea01ee81b6e8ab3d9534d76747","subject":"Merge \"ruff: Enable LOG checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/20b6241fef196fea01ee81b6e8ab3d9534d76747"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:44:36.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-07-13 19:32:07.000000000","tz":-180},"subject":"Make each dynamic pollster definition have its own discovery instance","message":"Make each dynamic pollster definition have its own discovery instance\n\nWhen coordination is enabled in Ceilometer, only one control node will\nexecute the pollster (either the hardcoded one or the dynamic pollster).\nThe synchronization is performed using the \"pooling.yml\" configuration\nin the \"ceilometer.polling.manager.PollingTask._internal_pollster_run\"\nmethod using the pollster name.\n\nHowever, the check (and further pollster execution) is only executed if\nthe code finds \"polling_resources\", which are the return of the\ndiscovery process. The discovery process also has a lock/coordination\nprocess inside it, and it uses the group of the discovery, which in the\ncase of the endpoint and Barbican discovery process, is a hardcoded\nstring called \"global\". This, in turn, locks all the execution of the\ndiscovery mechanism to the very first control node that executes it.\nAlso, it makes all pollsters that depend/need these discoveries be\nexecuted in the same control node.\n\nThis patch proposes a new structure, so each dynamic pollster has its\nown discovery instance. This, in turn, enables us to better distribute\nthe execution of the pollster in the cloud, as they are dependent on the\ndiscovery, which has a lock/coordination process as well.\n\nChange-Id: I86f657402dce2830cf1c2e438f7b4207b166bcf4\nSigned-off-by: Rafael Weingärtner \u003crafael@apache.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/fbac3d9266bedbb1cf1a58be916e9a9fcb3cc69b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/fbac3d9266bedbb1cf1a58be916e9a9fcb3cc69b"}]},"branch":"refs/heads/master"},"61de84b2dcf8ce1a6acbd9512ac18efba24ae934":{"kind":"REWORK","_number":5,"created":"2026-07-13 19:44:14.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/89/995389/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/89/995389/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/5"}}},"commit":{"parents":[{"commit":"20b6241fef196fea01ee81b6e8ab3d9534d76747","subject":"Merge \"ruff: Enable LOG checks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/20b6241fef196fea01ee81b6e8ab3d9534d76747"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:44:36.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-07-13 19:41:23.000000000","tz":-180},"subject":"Make each dynamic pollster definition have its own discovery instance","message":"Make each dynamic pollster definition have its own discovery instance\n\nWhen coordination is enabled in Ceilometer, only one control node will\nexecute the pollster (either the hardcoded one or the dynamic pollster).\nThe synchronization is performed using the \"pooling.yml\" configuration\nin the \"ceilometer.polling.manager.PollingTask._internal_pollster_run\"\nmethod using the pollster name.\n\nHowever, the check (and further pollster execution) is only executed if\nthe code finds \"polling_resources\", which are the return of the\ndiscovery process. The discovery process also has a lock/coordination\nprocess inside it, and it uses the group of the discovery, which in the\ncase of the endpoint and Barbican discovery process, is a hardcoded\nstring called \"global\". This, in turn, locks all the execution of the\ndiscovery mechanism to the very first control node that executes it.\nAlso, it makes all pollsters that depend/need these discoveries be\nexecuted in the same control node.\n\nThis patch proposes a new structure, so each dynamic pollster has its\nown discovery instance. This, in turn, enables us to better distribute\nthe execution of the pollster in the cloud, as they are dependent on the\ndiscovery, which has a lock/coordination process as well.\n\nChange-Id: I86f657402dce2830cf1c2e438f7b4207b166bcf4\nSigned-off-by: Rafael Weingärtner \u003crafael@apache.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/61de84b2dcf8ce1a6acbd9512ac18efba24ae934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/61de84b2dcf8ce1a6acbd9512ac18efba24ae934"}]},"branch":"refs/heads/master"},"8faf4d9149f17703866dd564c628af07fca6c784":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-07-14 11:58:51.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/89/995389/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/89/995389/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/6"}}},"commit":{"parents":[{"commit":"6593e47e08e06444cdeb279329ac07b92c42e9fe","subject":"ruff: Enable G (flake8-logging-format) checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/6593e47e08e06444cdeb279329ac07b92c42e9fe"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:44:36.000000000","tz":-180},"committer":{"name":"Rafael Weingartner","email":"rafael@apache.org","date":"2026-07-14 11:58:51.000000000","tz":0},"subject":"Make each dynamic pollster definition have its own discovery instance","message":"Make each dynamic pollster definition have its own discovery instance\n\nWhen coordination is enabled in Ceilometer, only one control node will\nexecute the pollster (either the hardcoded one or the dynamic pollster).\nThe synchronization is performed using the \"pooling.yml\" configuration\nin the \"ceilometer.polling.manager.PollingTask._internal_pollster_run\"\nmethod using the pollster name.\n\nHowever, the check (and further pollster execution) is only executed if\nthe code finds \"polling_resources\", which are the return of the\ndiscovery process. The discovery process also has a lock/coordination\nprocess inside it, and it uses the group of the discovery, which in the\ncase of the endpoint and Barbican discovery process, is a hardcoded\nstring called \"global\". This, in turn, locks all the execution of the\ndiscovery mechanism to the very first control node that executes it.\nAlso, it makes all pollsters that depend/need these discoveries be\nexecuted in the same control node.\n\nThis patch proposes a new structure, so each dynamic pollster has its\nown discovery instance. This, in turn, enables us to better distribute\nthe execution of the pollster in the cloud, as they are dependent on the\ndiscovery, which has a lock/coordination process as well.\n\nChange-Id: I86f657402dce2830cf1c2e438f7b4207b166bcf4\nSigned-off-by: Rafael Weingärtner \u003crafael@apache.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/8faf4d9149f17703866dd564c628af07fca6c784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/8faf4d9149f17703866dd564c628af07fca6c784"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"20b6241fef196fea01ee81b6e8ab3d9534d76747","ours":"61de84b2dcf8ce1a6acbd9512ac18efba24ae934","theirs":"6593e47e08e06444cdeb279329ac07b92c42e9fe","merge_strategy":"recursive","contains_conflicts":false}},"b31cf127349043d3885460649c2cb7da1b53419b":{"kind":"REWORK","_number":7,"created":"2026-07-14 13:51:48.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/89/995389/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/89/995389/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/7"}}},"commit":{"parents":[{"commit":"6593e47e08e06444cdeb279329ac07b92c42e9fe","subject":"ruff: Enable G (flake8-logging-format) checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/6593e47e08e06444cdeb279329ac07b92c42e9fe"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:44:36.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-07-14 13:51:27.000000000","tz":-180},"subject":"Make each dynamic pollster definition have its own discovery instance","message":"Make each dynamic pollster definition have its own discovery instance\n\nWhen coordination is enabled in Ceilometer, only one control node will\nexecute the pollster (either the hardcoded one or the dynamic pollster).\nThe synchronization is performed using the \"pooling.yml\" configuration\nin the \"ceilometer.polling.manager.PollingTask._internal_pollster_run\"\nmethod using the pollster name.\n\nHowever, the check (and further pollster execution) is only executed if\nthe code finds \"polling_resources\", which are the return of the\ndiscovery process. The discovery process also has a lock/coordination\nprocess inside it, and it uses the group of the discovery, which in the\ncase of the endpoint and Barbican discovery process, is a hardcoded\nstring called \"global\". This, in turn, locks all the execution of the\ndiscovery mechanism to the very first control node that executes it.\nAlso, it makes all pollsters that depend/need these discoveries be\nexecuted in the same control node.\n\nThis patch proposes a new structure, so each dynamic pollster has its\nown discovery instance. This, in turn, enables us to better distribute\nthe execution of the pollster in the cloud, as they are dependent on the\ndiscovery, which has a lock/coordination process as well.\n\nChange-Id: I86f657402dce2830cf1c2e438f7b4207b166bcf4\nSigned-off-by: Rafael Weingärtner \u003crafael@apache.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/b31cf127349043d3885460649c2cb7da1b53419b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/b31cf127349043d3885460649c2cb7da1b53419b"}]},"branch":"refs/heads/master"},"3b83d59fefbaf9075b579f0c07ec705c9a467292":{"kind":"REWORK","_number":8,"created":"2026-08-06 22:03:54.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/89/995389/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/89/995389/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/89/995389/8"}}},"commit":{"parents":[{"commit":"6593e47e08e06444cdeb279329ac07b92c42e9fe","subject":"ruff: Enable G (flake8-logging-format) checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/6593e47e08e06444cdeb279329ac07b92c42e9fe"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-06-29 17:44:36.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2026-08-06 22:03:38.000000000","tz":-180},"subject":"Make each dynamic pollster definition have its own discovery instance","message":"Make each dynamic pollster definition have its own discovery instance\n\nWhen coordination is enabled in Ceilometer, only one control node will\nexecute the pollster (either the hardcoded one or the dynamic pollster).\nThe synchronization is performed using the \"pooling.yml\" configuration\nin the \"ceilometer.polling.manager.PollingTask._internal_pollster_run\"\nmethod using the pollster name.\n\nHowever, the check (and further pollster execution) is only executed if\nthe code finds \"polling_resources\", which are the return of the\ndiscovery process. The discovery process also has a lock/coordination\nprocess inside it, and it uses the group of the discovery, which in the\ncase of the endpoint and Barbican discovery process, is a hardcoded\nstring called \"global\". This, in turn, locks all the execution of the\ndiscovery mechanism to the very first control node that executes it.\nAlso, it makes all pollsters that depend/need these discoveries be\nexecuted in the same control node.\n\nThis patch proposes a new structure, so each dynamic pollster has its\nown discovery instance. This, in turn, enables us to better distribute\nthe execution of the pollster in the cloud, as they are dependent on the\ndiscovery, which has a lock/coordination process as well.\n\nChange-Id: I86f657402dce2830cf1c2e438f7b4207b166bcf4\nSigned-off-by: Rafael Weingärtner \u003crafael@apache.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/3b83d59fefbaf9075b579f0c07ec705c9a467292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/3b83d59fefbaf9075b579f0c07ec705c9a467292"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
