)]}'
{"id":"openstack%2Fceilometer~998155","triplet_id":"openstack%2Fceilometer~master~Idaaf524db72bbce45e771d99d68b3608fb22afb0","project":"openstack/ceilometer","branch":"master","attention_set":{},"removed_from_attention_set":{"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2026-08-07 08:36:05.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-06 08:12:15.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2026-08-07 09:50:38.000000000","reason":"Change was submitted"},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2026-08-07 09:50:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idaaf524db72bbce45e771d99d68b3608fb22afb0","subject":"Log error on alarm and rgw pollster issues","status":"MERGED","created":"2026-07-21 14:22:41.000000000","updated":"2026-08-07 09:52:31.000000000","submitted":"2026-08-07 09:50:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998155","meta_rev_id":"cde6b1dd75a20710b06b77b331dc36700711f3a6","_number":998155,"virtual_id_number":998155,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-07 09:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":2,"date":"2026-08-07 08:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":2,"date":"2026-08-06 08:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":1,"date":"2026-08-07 08:36:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-21 15:31:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-22 14:26:38.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2026-07-23 03:38:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-07-24 15:23:58.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2026-08-06 03:02:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"32501bb336e8153d96b93c8eb3ddd1d84d190203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-21 14:22:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c27614ec0dabd941e38f1e1878cd91fd75a01e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 15:31:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0738170853924b918e9f2064b7a3a3a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f7cf1178e2240e7a7db96a9c74056a5 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4874a7c587f645649e9faf3cdfd3fc04 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a47e175532e473f82196a294e71e90e : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8641b89edc4c4111a969673dc3edbdc9 : SUCCESS in 5m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/63efd0742d53433191775ef0b45a11ef : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbfcff6c645045fb8d65746b079bb2db : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/730d6a12fb4f4f748befd4afe28676fd : SUCCESS in 6m 16s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/25fc550b75094aee86c4ae79eb32109d : SUCCESS in 57m 25s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/a98fd8549dcc45829847deffd34f0efd : SUCCESS in 1h 02m 11s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/656868844e0a4a7fbbf6958044030cc6 : SUCCESS in 47m 10s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/ca9f16d78c564e0eb82a69350959740d : SUCCESS in 56m 26s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/c5901427b16e48818ce04f05b5d5d658 : SUCCESS in 39m 37s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/7e82c362afaf4ac2943d6646ed4f7cda : SUCCESS in 33m 42s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/00baef8c50ae436a82e08a8a55f43b98 : SUCCESS in 44m 58s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"1f4d2f8d9473b3893c6d8ae890942d7c6fe885ca","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-07-22 14:26:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c49a3010bb551eef7e73b098813620b6273fc703","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-23 03:38:22.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9ec80f4ec70b02639e0a468e4caf32de1c0b8a82","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-23 07:28:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ec3e854dc62b2d07aa19473ebe35646d06e0953","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-23 18:24:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"36efb8e5928270aae48816883efa8a9890a1f717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-23 18:27:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f519b0ccbe73742a8f415a0d6d3ef087dc6b9430","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-23 18:27:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"074318c254883d71fe52740304ce91e7efcde590","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 19:40:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35a702b6774d45978b383a082d12713e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd4b6b1082b14010a1bdf72f56a9dcf2 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3917dbfc34424c976f854a9a04f5fd : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34ec86a70b3948b6a8457c5c71a4c79d : SUCCESS in 6m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2544c1a024b4bf389b56aff1bdc6d55 : SUCCESS in 3m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/326775c1a49a4a5e81aef159eb19846c : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0976a673fa8145b1be0010305de9c60c : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fc251b28d4b43009d729ee05d5840ba : SUCCESS in 3m 36s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/4845c7f423aa4ec7b24efa46ac380f91 : SUCCESS in 37m 07s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/09144ca0642b4822a65ed52b7c810b7d : SUCCESS in 57m 15s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/c7d551a6f46b425a94ad83bca32889f3 : SUCCESS in 1h 07m 47s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1e8b8d678df94061842a7d0094fa5e76 : SUCCESS in 52m 39s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/53d5d60adf0f4788abb6aa79f340e836 : SUCCESS in 44m 11s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1356515f94ca413585ae1083748db367 : SUCCESS in 51m 10s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/9fb4f98dac534983873010cd414703a2 : SUCCESS in 29m 55s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"d6723d4227c0fde3b1cc0c39d29c4768ca05c597","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-24 15:23:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"71470c22607d7cc34845dbb5306148e8ffafa922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-27 08:07:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"71ea1bf2f082f70ca8101fcb176c75b581ffe7a3","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-27 08:14:23.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"669518578aadb90f9915f3c1d4d8c5333855d16c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 09:21:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c969e3585ea40a9bfa47498633962ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60c64e699a4d4220b02a614d4081ba50 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79be060dd1e047199fb787ad7cef5f5e : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/944cffb4a7e14dd0a3279ae3d5af94fd : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a098d023ac384d5a85c0fac5b6b6e40d : SUCCESS in 5m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b65a44a2549047529b6567ae9e94ff0f : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/970c29ce21574aa08a227e8eef54eb96 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73b0f33b26124dbeb375222efef0fea3 : SUCCESS in 6m 03s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/0fc14c9adbd94bf68287cebbbfc6750f : SUCCESS in 1h 03m 20s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/abbc18301d65472da05f9716b35abe14 : SUCCESS in 57m 49s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/9ae497852dd94de6a04d0cbcbc928467 : SUCCESS in 1h 03m 13s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/fcde2a1258e649e6a8d7c955bab1e340 : SUCCESS in 50m 28s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/c2cab9a4a98a4523b82779577b09bf1f : SUCCESS in 1h 10m 57s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/13c9644599734f8d8a0da0bec23cb10e : SUCCESS in 45m 22s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/b81126e4abe84c2083366249fb371081 : SUCCESS in 25m 29s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"27f3c02c1c6d87b9655ccaa8ae3bf8bb9b73f8d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-27 12:32:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"89e5e8ecf9694afb029db6fbf26288c0b8125854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 13:48:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1390d52ffbef4ba98546ab557e27e9a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37e49fc4c9914f37a7b5c363c40c3b8b : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994de04d620f4e69acd893212ef66702 : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3e05efb3cd44566bd98f9903d43144f : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4aeec03cf46a48edaff8efd1487b7603 : SUCCESS in 5m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e902b0e5bc034420bb74876910c35d4a : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d2757717804f15b2ee40c6433ab25f : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/643746cf93414827bb717e76a0d4b625 : SUCCESS in 6m 57s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/730ba5f64481476390886fe3078087ac : SUCCESS in 1h 09m 38s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/dc2014079f1a486394d7251cda0f2571 : SUCCESS in 1h 07m 51s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/aabffd0025fa4d7b81a377f0c90372a3 : SUCCESS in 55m 26s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0818c436e7114537aa8c9dbcbcb042c1 : SUCCESS in 54m 30s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/41dedc5e50ab4b43b11f984df586a1e8 : SUCCESS in 40m 49s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/bfb13f3a7d49446085232ee948184e63 : SUCCESS in 50m 14s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/f10b816db7564126aa31320f8eed9ded : SUCCESS in 46m 37s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"5bb61464cf25f2b518e7d3a2a50e66cc79cb85b1","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-27 17:53:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0563239d49fd1c43a3acc5beca1249443e238ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-29 08:26:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7264a3dee83c215ec267926c367ca19c005749be","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-29 08:26:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d8d613fba1a9c3fa82c95e3232edb55b793e7230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 09:44:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5405d98116764135ab0d466d78cbbb61\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e836ec91481248d681d62b4c0945df09 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab23bfead5fa438a81aa6aa0c8c14434 : FAILURE in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e0f1ecc3f7d4c96bde3c9edd7260360 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb35fe3be80f452f9d882ad1dba08b9b : SUCCESS in 5m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f3b483491f8140af85c58a14746119fe : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51d30eb540b4ce7bcea926e98a52c8d : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6c01462060b4f1489ca6f802233a373 : SUCCESS in 5m 28s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/b4d12896671847c1920f402a3138d8c9 : SUCCESS in 1h 15m 57s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/ceef46bcec6a4e1ea031e322efac0c63 : SUCCESS in 1h 11m 22s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/76e1ca1e6d61441f934bb517f51c662d : SUCCESS in 1h 12m 42s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f5612be47f2044df9641951e6cc30414 : SUCCESS in 51m 08s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/de9f05666a894adc8a92b449dcf3c9ed : SUCCESS in 45m 18s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/aeffb44605804e36ad2de0892f66021b : SUCCESS in 52m 24s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/f77c4e6b6b68499f8d1abaf376ee93bc : SUCCESS in 45m 44s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5},{"id":"db249430acdcf5660c3fa3acbfc984d057b6634a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-07-29 10:09:15.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"5ab8c2454103d19aabf35b879df71e1b35e12668","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 11:29:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8b26a19132e45bda74e9e3647f938e0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e98b212f8b0c4b74b057e82f544f6898 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da7e353cd16749edbc33e0be9c03fb3d : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef20659970804cb88a6bed6a15fa82cf : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/361bd1bd9c00475bb2d3eed3162b295e : SUCCESS in 6m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5cabbc0e57074574b11a54bfaa8a39dd : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1de3827d221e4200b248aaeaa8383296 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb236e67da1e4b0fa02a3229063f2ff8 : SUCCESS in 6m 30s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/e53e0c0246ef47b4b2e7a66b181eeaa3 : SUCCESS in 1h 09m 41s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/e279041157e2496196a91fea401e6282 : SUCCESS in 1h 04m 44s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/f0cec7b6bca04ca9a239700d629a9501 : SUCCESS in 48m 52s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2c06d522a7024907a584dad1f8750bbb : SUCCESS in 49m 07s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/bb4ad6cf5076498c851ae8014ffdb30c : SUCCESS in 44m 42s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/029753fff24a4cc48074e0c9dadc9b7d : SUCCESS in 53m 14s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/a2a3bdf39d854db3aca4eb330db81d1d : SUCCESS in 47m 59s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":6},{"id":"79dee6adf59fd51ce76eb4628501b6d8f02189bd","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-05 17:16:46.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"5eee37aa0685fc20dc99c41ac68a6812d4ef97df","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-06 03:02:50.000000000","message":"Patch Set 6: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"be3e46f8099ed829c6053e0528058c8c4e4b8ca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-08-06 07:13:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"86a8ae7c3d7f59b2e0d85f9497e771060f7fe5b6","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-08-06 07:14:38.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7a4e6a4919c19780071bce1ef0203205216ec9bd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-06 08:12:15.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"aa74b14f9d741e3496a4d7ae6cf95b1e7ef7f27a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 08:27:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b907222f75b44aeaf591116b790ebb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0a2d80d45aa43bb8ead35208c2653b2 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34f3e5e473e94244baac14228359c414 : FAILURE in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce3158838f80463d8d27e9d7afb987ae : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9928cb62ec81461a81a9dc092226e1ad : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a471f05f68304f9498c4ac417abf7f1e : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81f71857def412d8e9446d135724c01 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7417b60e3b34763955529a493ac9554 : SUCCESS in 2m 59s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/996a60bd403e4c1aa959f178724d0f4e : SUCCESS in 1h 08m 20s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/c5435630808a4f9b97bf2e4bfa981df1 : SUCCESS in 1h 08m 10s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/b29e2759f5a04747b78ec599c6a1f425 : FAILURE in 47m 45s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2780125a07af430796a17742ee09c0d3 : SUCCESS in 1h 11m 44s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/bc274e7ad6e544b39f6df5beb157e895 : SUCCESS in 39m 55s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f63d43dc87da41a4970f198e7f28558b : FAILURE in 30m 02s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/2545ae6b51a042a584deba4419321419 : SUCCESS in 43m 18s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"3b56bcbfb9cff6ab89b7e73e9de0018ab2b65334","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-08-06 21:07:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"72d97b84eec90229aec2718363d149a730ba5c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 22:30:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/281c527c592a436fb6470f44aafc868b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/773b5d24f76b48b2865627aefca5017c : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebf7df546fb4190b5a972a322a5ff0e : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7119c866ea7149c4b285075429b26997 : SUCCESS in 6m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c2952b4df9b4e8f92de289b6b468ffc : SUCCESS in 6m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34b01ccc1d5749f9967e679daef1d886 : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497b62a77b614d05a1dcff93af39c749 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/807b26bbc5364d74a6a162f55b5264ec : SUCCESS in 6m 20s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/d9254d48275940989993c7ee2cd3eb8f : SUCCESS in 1h 16m 46s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/2e052b703caa4448ad54e4cc675cc86e : SUCCESS in 59m 39s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/fa2e238a7d014064ad97eccbfebb6742 : SUCCESS in 1h 07m 27s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8f43db3e04d84543912e34c1754d7be0 : SUCCESS in 55m 43s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/d8e28dab04e944e1946d92b623e7941f : SUCCESS in 50m 50s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d4a16d9204a74915bbbc72dcf4a4c3fc : SUCCESS in 52m 54s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/21547d0e21f542babbf9b515968004fc : SUCCESS in 22m 30s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":7},{"id":"a60d4d3b4d879a572efd2ab0ccbbf2fba8043e71","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2026-08-07 08:36:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"5060b5394cc50cf1ef9ed9ad7acf1548cb26bab9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 08:36:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e6a7b39a25b5c1d0bfafae52a2ef2c2a0b0ee281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 09:50:38.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5a3c3174b0d42eb97019d9cb83c26a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c962cc4e9da4866bc7df876b10b90fc : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/faa1321a17ec4886a7aded22859f9702 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/55020e50bbd143f6be13ed05de489c49 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fc26a87fe844e9c9b85813576dd5c50 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ed075cbd5b44b18a1a6094961d0fd2e : SUCCESS in 2m 39s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/57f9dcaf7cc14cbbbb28d735b7acc8d0 : SUCCESS in 1h 13m 54s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d5eaed5d3b0542f792b60e6b990d2f92 : SUCCESS in 53m 20s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/16d7200bfaa34ce5be6b2b757b2b3379 : SUCCESS in 42m 29s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/83245e526d224bf98abc6b837eb289fe : SUCCESS in 56m 17s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/31ea1074a8d249d88966d8facd27bb8e : SUCCESS in 43m 18s","accounts_in_message":[],"_revision_number":7},{"id":"f3ea3553848407ebec2821d6fb40c7b78a1a1ddc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 09:50:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"cde6b1dd75a20710b06b77b331dc36700711f3a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 09:52:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f1f366a11474ac8b9d5c7580bcb70f0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6d5e7076a9443693ee2f42098975a1 : SUCCESS in 1m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d379af4fe2104a6ebbe5aa633cac34b2 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0c5e37de9e348df3f8ba9d13d846a9626450eb5f","revisions":{"2ec76805b73e28ba6749a743bd61a3080eec891b":{"kind":"REWORK","_number":1,"created":"2026-07-21 14:22:41.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/55/998155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/55/998155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/1"}}},"commit":{"parents":[{"commit":"1a71ec923663adb8808354d4c1e839b1dc25feae","subject":"Merge \"Fix flavor_id mapping for event-driven instance resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/1a71ec923663adb8808354d4c1e839b1dc25feae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-21 14:12:35.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-21 14:21:01.000000000","tz":-240},"subject":"Log error on alarm and rgw pollster issues","message":"Log error on alarm and rgw pollster issues\n\nWhen an issue is encountered in the alarm or rgw pollsters, log error\ninstead of throwing an exception with the whole traceback. This follows\nthe example of other pollsters (like manila or designate), which just\nlog an error line about skipping that particullar pollster. This\nsignificantly improves the logs especially in situations where all\npollsters are enabled at once with the \"*\" wildcard.\n\nGenerated-by: Claude-Code claude-opus-4-6\nCloses-Bug: #2161437\nChange-Id: Idaaf524db72bbce45e771d99d68b3608fb22afb0\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/2ec76805b73e28ba6749a743bd61a3080eec891b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/2ec76805b73e28ba6749a743bd61a3080eec891b"}]},"branch":"refs/heads/master"},"29099e706d0e92f041e65293c775ee8e5df61831":{"kind":"REWORK","_number":2,"created":"2026-07-23 18:27:11.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/55/998155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/55/998155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/2"}}},"commit":{"parents":[{"commit":"1a71ec923663adb8808354d4c1e839b1dc25feae","subject":"Merge \"Fix flavor_id mapping for event-driven instance resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/1a71ec923663adb8808354d4c1e839b1dc25feae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-21 14:12:35.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-23 18:26:50.000000000","tz":-240},"subject":"Log error on alarm and rgw pollster issues","message":"Log error on alarm and rgw pollster issues\n\nWhen an issue is encountered in the alarm or rgw pollsters, log error\ninstead of throwing an exception with the whole traceback. This follows\nthe example of other pollsters (like manila or designate), which just\nlog an error line about skipping that particullar pollster. This\nsignificantly improves the logs especially in situations where all\npollsters are enabled at once with the \"*\" wildcard.\n\nGenerated-by: Claude-Code claude-opus-4-6\nCloses-Bug: #2161437\nChange-Id: Idaaf524db72bbce45e771d99d68b3608fb22afb0\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/29099e706d0e92f041e65293c775ee8e5df61831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/29099e706d0e92f041e65293c775ee8e5df61831"}]},"branch":"refs/heads/master"},"23225c74909e39b45346cc419b41170548359b4f":{"kind":"REWORK","_number":3,"created":"2026-07-27 08:07:41.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/55/998155/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/55/998155/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/3"}}},"commit":{"parents":[{"commit":"1a71ec923663adb8808354d4c1e839b1dc25feae","subject":"Merge \"Fix flavor_id mapping for event-driven instance resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/1a71ec923663adb8808354d4c1e839b1dc25feae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-21 14:12:35.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-27 07:59:03.000000000","tz":0},"subject":"Log error on alarm and rgw pollster issues","message":"Log error on alarm and rgw pollster issues\n\nWhen an issue is encountered in the alarm or rgw pollsters, log error\ninstead of throwing an exception with the whole traceback. This follows\nthe example of other pollsters (like manila or designate), which just\nlog an error line about skipping that particullar pollster. This\nsignificantly improves the logs especially in situations where all\npollsters are enabled at once with the \"*\" wildcard.\n\nGenerated-by: Claude-Code claude-opus-4-6\nCloses-Bug: #2161437\nChange-Id: Idaaf524db72bbce45e771d99d68b3608fb22afb0\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/23225c74909e39b45346cc419b41170548359b4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/23225c74909e39b45346cc419b41170548359b4f"}]},"branch":"refs/heads/master"},"ce2733cd2555e346af0d42a0779c90a4530310e9":{"kind":"REWORK","_number":4,"created":"2026-07-27 12:32:25.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/55/998155/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/55/998155/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/4"}}},"commit":{"parents":[{"commit":"1a71ec923663adb8808354d4c1e839b1dc25feae","subject":"Merge \"Fix flavor_id mapping for event-driven instance resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/1a71ec923663adb8808354d4c1e839b1dc25feae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-21 14:12:35.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-27 12:28:58.000000000","tz":0},"subject":"Log error on alarm and rgw pollster issues","message":"Log error on alarm and rgw pollster issues\n\nWhen an issue is encountered in the alarm or rgw pollsters, log error\ninstead of throwing an exception with the whole traceback. This follows\nthe example of other pollsters (like manila or designate), which just\nlog an error line about skipping that particullar pollster. This\nsignificantly improves the logs especially in situations where all\npollsters are enabled at once with the \"*\" wildcard.\n\nGenerated-by: Claude-Code claude-opus-4-6\nCloses-Bug: #2161437\nChange-Id: Idaaf524db72bbce45e771d99d68b3608fb22afb0\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/ce2733cd2555e346af0d42a0779c90a4530310e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/ce2733cd2555e346af0d42a0779c90a4530310e9"}]},"branch":"refs/heads/master"},"8f0556aa8a928269e4af31f2dc0b1c52726386ce":{"kind":"REWORK","_number":5,"created":"2026-07-29 08:26:11.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/55/998155/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/55/998155/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/5"}}},"commit":{"parents":[{"commit":"1a71ec923663adb8808354d4c1e839b1dc25feae","subject":"Merge \"Fix flavor_id mapping for event-driven instance resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/1a71ec923663adb8808354d4c1e839b1dc25feae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-21 14:12:35.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-29 08:22:15.000000000","tz":0},"subject":"Log error on alarm and rgw pollster issues","message":"Log error on alarm and rgw pollster issues\n\nWhen an issue is encountered in the alarm or rgw pollsters, log error\ninstead of throwing an exception with the whole traceback. This follows\nthe example of other pollsters (like manila or designate), which just\nlog an error line about skipping that particullar pollster. This\nsignificantly improves the logs especially in situations where all\npollsters are enabled at once with the \"*\" wildcard.\n\nGenerated-by: Claude-Code claude-opus-4-6\nCloses-Bug: #2161437\nChange-Id: Idaaf524db72bbce45e771d99d68b3608fb22afb0\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/8f0556aa8a928269e4af31f2dc0b1c52726386ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/8f0556aa8a928269e4af31f2dc0b1c52726386ce"}]},"branch":"refs/heads/master"},"ee54123c363a1be730e3e5d4610a04fde39ba12a":{"kind":"REWORK","_number":6,"created":"2026-07-29 10:09:15.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/55/998155/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/55/998155/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/6"}}},"commit":{"parents":[{"commit":"1a71ec923663adb8808354d4c1e839b1dc25feae","subject":"Merge \"Fix flavor_id mapping for event-driven instance resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/1a71ec923663adb8808354d4c1e839b1dc25feae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-21 14:12:35.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-29 10:05:16.000000000","tz":0},"subject":"Log error on alarm and rgw pollster issues","message":"Log error on alarm and rgw pollster issues\n\nWhen an issue is encountered in the alarm or rgw pollsters, log error\ninstead of throwing an exception with the whole traceback. This follows\nthe example of other pollsters (like manila or designate), which just\nlog an error line about skipping that particullar pollster. This\nsignificantly improves the logs especially in situations where all\npollsters are enabled at once with the \"*\" wildcard.\n\nGenerated-by: Claude-Code claude-opus-4-6\nCloses-Bug: #2161437\nChange-Id: Idaaf524db72bbce45e771d99d68b3608fb22afb0\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/ee54123c363a1be730e3e5d4610a04fde39ba12a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/ee54123c363a1be730e3e5d4610a04fde39ba12a"}]},"branch":"refs/heads/master"},"0c5e37de9e348df3f8ba9d13d846a9626450eb5f":{"kind":"REWORK","_number":7,"created":"2026-08-06 07:13:39.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/55/998155/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/55/998155/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/55/998155/7"}}},"commit":{"parents":[{"commit":"1a71ec923663adb8808354d4c1e839b1dc25feae","subject":"Merge \"Fix flavor_id mapping for event-driven instance resources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/1a71ec923663adb8808354d4c1e839b1dc25feae"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-07-21 14:12:35.000000000","tz":0},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2026-08-06 07:08:39.000000000","tz":0},"subject":"Log error on alarm and rgw pollster issues","message":"Log error on alarm and rgw pollster issues\n\nWhen an issue is encountered in the alarm or rgw pollsters, log error\ninstead of throwing an exception with the whole traceback. This follows\nthe example of other pollsters (like manila or designate), which just\nlog an error line about skipping that particullar pollster. This\nsignificantly improves the logs especially in situations where all\npollsters are enabled at once with the \"*\" wildcard.\n\nGenerated-by: Claude-Code claude-opus-4-6\nCloses-Bug: #2161437\nChange-Id: Idaaf524db72bbce45e771d99d68b3608fb22afb0\nSigned-off-by: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/0c5e37de9e348df3f8ba9d13d846a9626450eb5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/0c5e37de9e348df3f8ba9d13d846a9626450eb5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
