)]}'
{"id":"openstack%2Fceilometer~998851","triplet_id":"openstack%2Fceilometer~master~I8bb55b32eb88aa81d9362eac112d7beb44b03d3d","project":"openstack/ceilometer","branch":"master","topic":"ceilometer_use_openstack_sdk","attention_set":{"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2026-08-04 22:18:05.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8bb55b32eb88aa81d9362eac112d7beb44b03d3d","subject":"Use openstacksdk in ceilometer/novaclient","status":"NEW","created":"2026-07-27 20:53:52.000000000","updated":"2026-08-13 13:51:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aa55bd65153f79647f3718fc7ee5f3c761a40174","_number":998851,"virtual_id_number":998851,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-13 13:51:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-27 22:05:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"966d2e70364171caab7c8ea7df580fd2540ff647","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-27 20:53:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca8bba49a9449c7677a8005e83aaba2a35274c3b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-07-27 20:57:39.000000000","message":"Topic set to ceilometer_use_openstack_sdk","accounts_in_message":[],"_revision_number":1},{"id":"83fe861b1a05003fd335d7751d2ecf139897b939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 22:05:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57689ee6a1d94e6cba7f358be31f8c5d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44f8d9b168b942faa25fdbb4e505a163 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29cda2659f84b4894d5400771afcc2a : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95d24d105472418fac829fe4d0535ac9 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2915899f530d4dae97f17a9e2ccc8e2e : SUCCESS in 5m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6b1146951c344b19884a53b7f8153a4 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a934b49ad52466e9da455cf9bf675aa : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a75fea1b78734c059ae09417fe74b46b : SUCCESS in 6m 18s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/dd6de60691294bb9b1fbeaf881684fc8 : SUCCESS in 1h 08m 28s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/252c49ffa5554e8bbb121a9b69fcf13d : SUCCESS in 56m 37s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/d03b822f4b154190a700485ac59c3d2f : SUCCESS in 1h 07m 41s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bb7c120a78fa43888f278434f1dd4b3b : SUCCESS in 54m 18s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/4a37bf1d6d214ba3803848ecebb6805a : SUCCESS in 41m 31s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/36f078663ea84b1dbd6113b8c92ac1a5 : SUCCESS in 39m 19s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/d450f689c33f497a86ed8ee43bbb56e3 : SUCCESS in 43m 19s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"34e7624e07300e5259d3f6ca89b23dfb0241e77d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-04 20:58:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c2619f74d76a502b18a5ce55c9d43cc223acb881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 22:18:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa2ae63f638545219b01ea0eed0dee7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7103fd0e2e3455f97bb316a75d89d83 : FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7219628164cd490db24608a3c4842640 : FAILURE in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/949e48847c584069a424797de15867b4 : FAILURE in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/306f0a5e39be4b0a8720aecb477c6fef : FAILURE in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ceadd868dac84aeab593dcf89cbbd93a : FAILURE in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea8f5b95d1dc4aecb3344973089a446c : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c7a05e1cb8d4533aacdf4d211879e21 : SUCCESS in 6m 28s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/7c86ace8bbaa4f849ef1f599e31dd16d : SUCCESS in 1h 13m 10s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/896dca9485ef4052ad1288062d57d3d3 : SUCCESS in 1h 00m 37s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/589376f7c85948159dcb9c05dd97f7d0 : SUCCESS in 1h 02m 31s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3dd71f350e1249ee89a9e032bfdf1919 : SUCCESS in 39m 55s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/9023f589c2824c8baa4b0669d9e41f73 : SUCCESS in 45m 50s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/df006e38a2394a63a19bc3cdbee8c2a8 : SUCCESS in 52m 19s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/6a56c054b0d14f4ab3a41f7ab859c7ca : SUCCESS in 42m 33s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"e0fc54aa663148686a0a273e044f9059445266c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-06 20:36:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"71837235347665d28f06b34c3e671c42d3df20a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 21:46:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab81405677d94cf88aa52278ae6e9e05\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2cbffaae0c04bdbbec30922af81c34d : FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b7b51ee0d3b4059be5b185d0d343086 : FAILURE in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/febdea5d03294b53a68c429d44425128 : FAILURE in 2m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd1c9576be47498192258f2b9043be44 : FAILURE in 2m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/40f55496031948e2a6a96d18ed536811 : FAILURE in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b5d99455be43ffb27b9764304ecd1b : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05fd4a30876648d9bdbbd56f8c36432f : SUCCESS in 3m 18s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/c1eb521ef2c44f44ae83eca29702cf98 : SUCCESS in 45m 18s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/683e3204f7264adebdcfe4e027ec81ca : SUCCESS in 1h 01m 10s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/11fbfc8bdaff43f2a1b026a42f89c9a7 : SUCCESS in 1h 03m 08s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/725c19dabb1c47dbb228613735911f2f : SUCCESS in 51m 19s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/52953e5d80264c629d21520b4f838573 : SUCCESS in 42m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c1725e39143b451784580644d9edb245 : SUCCESS in 43m 26s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/918eac7be31a4041b2c28e80f7ca29d5 : SUCCESS in 47m 08s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"32f52f77e50f753bddf84563e1c19feca73dacd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-10 18:11:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"619f3ac773bc327cd8f6960f021ddbcb369b41c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 19:41:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38346b89105245deb5547a4026efe85f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/558b999b5af0464083b13f5db4b8718b : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6353d4fd4b7e470b8c073aa835844d9f : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d61504622d2342d99f23a0996bf43e9a : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/005a09aa5c254e19b269930a53482d5b : SUCCESS in 5m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7eef52add69b480eb56233a1dce6c571 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13473e77fc554d689d7edf0e5f9b3992 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a69fa21422c4dab9ab4abd9d6f52299 : SUCCESS in 3m 40s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/9561dad171d74c80a6a9cc8bf88ba51f : SUCCESS in 1h 22m 04s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/b26153fe8270476e985a63e3b560c740 : SUCCESS in 1h 11m 40s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/8f822eedd514461883ab7aa1567aea5c : SUCCESS in 59m 07s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f4a1757bdc9040e59aa81f005b47b0d5 : SUCCESS in 53m 16s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/beb6c827aaff48afb95cd94bc1047dd6 : SUCCESS in 45m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/5548e1f3cec14e3daf63ca3bd9417ff1 : SUCCESS in 49m 02s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/e639abe094b349f79de993db3d2fde47 : SUCCESS in 45m 39s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":4},{"id":"0da1af321af3d91a1467203e8c9e55ce98d879ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-08-13 12:23:33.000000000","message":"Uploaded patch set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"aa55bd65153f79647f3718fc7ee5f3c761a40174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 13:51:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcfeb4800b164724bfa05161b1fb192e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac44bf6709e24fd5b5bb3d102b055aea : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e53aa0ba80a04d0d93602c7a78c566cf : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8553c8bf62442c497e6eb67f3a101f8 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78b25591cfcb4d8d90409dc19637e32b : SUCCESS in 4m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f7a7360330734b9880ad07064db729e9 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0c0833a4704b0b823c33fae691d60c : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be6822a517164d32aed2ca9400def90f : SUCCESS in 6m 16s\n- grenade-ceilometer https://zuul.opendev.org/t/openstack/build/321233134c6444e8bd1c721606d5bdf2 : FAILURE in 9m 23s\n- telemetry-dsvm-integration-centos-9s https://zuul.opendev.org/t/openstack/build/24df4a7c222c49ea83c65cccc64453ad : SUCCESS in 1h 20m 33s (non-voting)\n- telemetry-dsvm-integration-centos-9s-fips https://zuul.opendev.org/t/openstack/build/4f4551fb73c342d09e57aaad0173a021 : FAILURE in 25m 52s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c6541764bfc94b2493060913a36f28ce : FAILURE in 13m 01s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/80320dfc211d4fc28ad4fb13850dda9e : SUCCESS in 47m 46s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a3e61fb397054543b027c0ca6b9c191d : SUCCESS in 1h 12m 28s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/2a2bb95c69ee4db6be8177565035e71e : FAILURE in 9m 00s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job telemetry-dsvm-integration-centos-9s-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"aa726e09fd9d016b29b91dc2f0731b24ebad3868","revisions":{"aa2d25bc11f8ff316db652b8144a615e44d0e081":{"kind":"REWORK","_number":1,"created":"2026-07-27 20:53:52.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/51/998851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/51/998851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/1"}}},"commit":{"parents":[{"commit":"a2b13e7cd9180fa193961df61e34bdedc00fa812","subject":"[WIP] Add a Fake SDK Compute client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/a2b13e7cd9180fa193961df61e34bdedc00fa812"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-24 20:49:27.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-27 20:28:42.000000000","tz":-240},"subject":"[WIP] Use openstacksdk in ceilometer/novaclient","message":"[WIP] Use openstacksdk in ceilometer/novaclient\n\nChange-Id: I8bb55b32eb88aa81d9362eac112d7beb44b03d3d\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/aa2d25bc11f8ff316db652b8144a615e44d0e081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/aa2d25bc11f8ff316db652b8144a615e44d0e081"}]},"branch":"refs/heads/master"},"ac35c2a0c8063b9d736a23a87d2c5d0954f6b9a9":{"kind":"REWORK","_number":2,"created":"2026-08-04 20:58:17.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/51/998851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/51/998851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/2"}}},"commit":{"parents":[{"commit":"e38d47160ef0510df8c6f025cde1d9d0757db1a8","subject":"[WIP] Add a FakeSDKCompute test fixture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/e38d47160ef0510df8c6f025cde1d9d0757db1a8"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-24 20:49:27.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-08-04 20:57:16.000000000","tz":-240},"subject":"[WIP] Use openstacksdk in ceilometer/novaclient","message":"[WIP] Use openstacksdk in ceilometer/novaclient\n\nChange-Id: I8bb55b32eb88aa81d9362eac112d7beb44b03d3d\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/ac35c2a0c8063b9d736a23a87d2c5d0954f6b9a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/ac35c2a0c8063b9d736a23a87d2c5d0954f6b9a9"}]},"branch":"refs/heads/master"},"54d631ffbb71dc8e99c397ad1197b48c5cf014c3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-08-06 20:36:54.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/51/998851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/51/998851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/3"}}},"commit":{"parents":[{"commit":"78443cce4e458eb066aa640578ae7514a07fd587","subject":"[WIP] Add a FakeSDKCompute test fixture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/78443cce4e458eb066aa640578ae7514a07fd587"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-24 20:49:27.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-08-06 20:36:10.000000000","tz":-240},"subject":"[WIP] Use openstacksdk in ceilometer/novaclient","message":"[WIP] Use openstacksdk in ceilometer/novaclient\n\nChange-Id: I8bb55b32eb88aa81d9362eac112d7beb44b03d3d\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/54d631ffbb71dc8e99c397ad1197b48c5cf014c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/54d631ffbb71dc8e99c397ad1197b48c5cf014c3"}]},"branch":"refs/heads/master"},"eff8c963e158b51a03892b2926b3c95b319dd7f5":{"kind":"REWORK","_number":4,"created":"2026-08-10 18:11:26.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/51/998851/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/51/998851/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/4"}}},"commit":{"parents":[{"commit":"3a39ff0f37e58e6e5b03b84c3aa9dfb4d3d3a03a","subject":"[WIP] Add a FakeSDKCompute test fixture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/3a39ff0f37e58e6e5b03b84c3aa9dfb4d3d3a03a"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-24 20:49:27.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-08-10 18:03:20.000000000","tz":-240},"subject":"[WIP] Use openstacksdk in ceilometer/novaclient","message":"[WIP] Use openstacksdk in ceilometer/novaclient\n\nChange-Id: I8bb55b32eb88aa81d9362eac112d7beb44b03d3d\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/eff8c963e158b51a03892b2926b3c95b319dd7f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/eff8c963e158b51a03892b2926b3c95b319dd7f5"}]},"branch":"refs/heads/master"},"aa726e09fd9d016b29b91dc2f0731b24ebad3868":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-08-13 12:23:33.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/51/998851/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ceilometer","ref":"refs/changes/51/998851/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ceilometer refs/changes/51/998851/5"}}},"commit":{"parents":[{"commit":"fa6f3656a5b3f67da6719bbd640ba8417e9b81ac","subject":"Add a FakeSDKCompute test fixture","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/fa6f3656a5b3f67da6719bbd640ba8417e9b81ac"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-07-24 20:49:27.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-08-13 12:23:17.000000000","tz":-240},"subject":"Use openstacksdk in ceilometer/novaclient","message":"Use openstacksdk in ceilometer/novaclient\n\nChange-Id: I8bb55b32eb88aa81d9362eac112d7beb44b03d3d\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/aa726e09fd9d016b29b91dc2f0731b24ebad3868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ceilometer/commit/aa726e09fd9d016b29b91dc2f0731b24ebad3868"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
