)]}'
{"id":"openstack%2Fcharm-ceilometer-agent~746051","triplet_id":"openstack%2Fcharm-ceilometer-agent~stable%2F20.08~I6667b5c6513e6b16ae39bdd65d4cc4376081d55d","project":"openstack/charm-ceilometer-agent","branch":"stable/20.08","topic":"release-20.08","hashtags":[],"change_id":"I6667b5c6513e6b16ae39bdd65d4cc4376081d55d","subject":"Updates for stable branch creation","status":"MERGED","created":"2020-08-13 10:28:03.000000000","updated":"2020-08-13 12:18:17.000000000","submitted":"2020-08-13 12:18:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746051-1597321097634-5080a3ed","meta_rev_id":"1c88fedf8decebf7711fc6a70cfde4dffe642e7a","_number":746051,"virtual_id_number":746051,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 12:18:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 12:07:59.000000000","_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-13 12:07:59.000000000","_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 12:07:59.000000000","updated_by":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"reviewer":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"state":"REVIEWER"},{"updated":"2020-08-13 12:18:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"161a9766c7182f7728955d8ea781aaa083e6d45c","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2020-08-13 10:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f992c27bfb7fe13b0665c5a8f65564dfd1dd918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 10:37:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239c7ebde94343de8b877116c23c6b0e : SUCCESS in 3m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3b095554ea384beeb922760b964188eb : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/351edc44c4294910bbc3146268abc863 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35c5fe793f244304ac64e7646697b917 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"04b2962875e888f9fadd6ab50bf7f708802dbe81","author":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"date":"2020-08-13 12:07:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"ce48a26a423893ed405e0581bbb8a7d9323154d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:08:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"326fdb6763a39b76ee5f640f5cda41f91db3afb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:18:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"015a0e2ffc32733527eb22751e4b331a52390e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:18:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b677ed2f6cb4c858d746770cae28fd8 : SUCCESS in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c662550efa1d4224acf52b24c0db7910 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddeda1cf0eb7428db77df43e02b7b6ca : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/901013fba28146e2a509d92551082a70 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74d6ed19ee292aa2ad946f4b049a981fd4974057","revisions":{"74d6ed19ee292aa2ad946f4b049a981fd4974057":{"kind":"REWORK","_number":1,"created":"2020-08-13 10:28:03.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/51/746051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-ceilometer-agent","ref":"refs/changes/51/746051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-ceilometer-agent refs/changes/51/746051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-ceilometer-agent refs/changes/51/746051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-ceilometer-agent refs/changes/51/746051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-ceilometer-agent refs/changes/51/746051/1"}}},"commit":{"parents":[{"commit":"1f7ddf529fb42134826e2786cf087f88ce6d7909","subject":"Release sync for 20.08","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-ceilometer-agent/commit/1f7ddf529fb42134826e2786cf087f88ce6d7909"}]}],"author":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:27:52.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:28:01.000000000","tz":60},"subject":"Updates for stable branch creation","message":"Updates for stable branch creation\n\nSet default branch for git review/gerrit.\n\nSwitch tests to stable.\n\nSwitch to using stable charm-helpers branch.\n\nSwitch to using stable charm.openstack branch.\n\nChange-Id: I6667b5c6513e6b16ae39bdd65d4cc4376081d55d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-ceilometer-agent/commit/74d6ed19ee292aa2ad946f4b049a981fd4974057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-ceilometer-agent/commit/74d6ed19ee292aa2ad946f4b049a981fd4974057"}]},"branch":"refs/heads/stable/20.08"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
