)]}'
{"id":"openstack%2Fcharm-ceph-proxy~746055","triplet_id":"openstack%2Fcharm-ceph-proxy~stable%2F20.08~I8aee823f98a9c4b3a56019b445410c645b2bb423","project":"openstack/charm-ceph-proxy","branch":"stable/20.08","topic":"release-20.08","hashtags":[],"change_id":"I8aee823f98a9c4b3a56019b445410c645b2bb423","subject":"Updates for stable branch creation","status":"MERGED","created":"2020-08-13 10:29:02.000000000","updated":"2020-08-13 12:00:12.000000000","submitted":"2020-08-13 12:00:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746055-1597320012446-5080a3ed","meta_rev_id":"4b9050e6a3ecdde32c4e60e217a97b2ad866450c","_number":746055,"virtual_id_number":746055,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 12:00:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 11:48:55.000000000","_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-13 11:48:55.000000000","_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 11:48:55.000000000","updated_by":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"reviewer":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"state":"REVIEWER"},{"updated":"2020-08-13 12:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcb70e570b1b07d7845d6f20bdefa1b6f4399453","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2020-08-13 10:29:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0330971762df1183e3541892a395c848a5f20a42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 10:40:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5f5bc5d3b1340d9a51d93a9b8cc3740 : SUCCESS in 3m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/45a584f0d4d64246937371a0ceb67c81 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/492cb3e0e10b41b7b17bfaaddd23834b : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/180e562c797f4111adea9c31597a8cda : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ff2bed23dfe33a843a71c153d1aa105721c72831","author":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"date":"2020-08-13 11:48:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"1d95d2af35c4b393157db1c2a01afa1ce7ccce12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 11:49:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef9d901a8945f8f1b9a3aa1e878258d82f6a4e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:00:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ca2c269515c4c619302042f4c906e31 : SUCCESS in 3m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8f48c028ff4845a9bdee58abc7fed94f : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d061cc749ddf4c4897165b9d77612d03 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d7b1df7f2334d1fbfd7a5b895fb8327 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0f498af3ff1d9984855867c9340850446554d2d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:00:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60260cc66cfbe55524033be5552494c926235ae9","revisions":{"60260cc66cfbe55524033be5552494c926235ae9":{"kind":"REWORK","_number":1,"created":"2020-08-13 10:29:02.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/55/746055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-ceph-proxy","ref":"refs/changes/55/746055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-ceph-proxy refs/changes/55/746055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-ceph-proxy refs/changes/55/746055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-ceph-proxy refs/changes/55/746055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-ceph-proxy refs/changes/55/746055/1"}}},"commit":{"parents":[{"commit":"8cc52727a20173f72963a28bef883fdfddb96cda","subject":"Fix actions section of the README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-ceph-proxy/commit/8cc52727a20173f72963a28bef883fdfddb96cda"}]}],"author":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:28:51.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:29:00.000000000","tz":60},"subject":"Updates for stable branch creation","message":"Updates for stable branch creation\n\nSet default branch for git review/gerrit.\n\nSwitch tests to stable.\n\nSwitch to using stable charm-helpers branch.\n\nSwitch to using stable charm.openstack branch.\n\nChange-Id: I8aee823f98a9c4b3a56019b445410c645b2bb423\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-ceph-proxy/commit/60260cc66cfbe55524033be5552494c926235ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-ceph-proxy/commit/60260cc66cfbe55524033be5552494c926235ae9"}]},"branch":"refs/heads/stable/20.08"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
