)]}'
{"/PATCHSET_LEVEL":[{"author":{"_account_id":31836,"name":"Facundo Ciccioli","email":"facundo.ciccioli@canonical.com","username":"facu"},"change_message_id":"1d1ee2492eff2f77174d9f4dbb7e7b48b7a67998","unresolved":true,"context_lines":[],"source_content_type":"","patch_set":7,"id":"ff1c16bc_7af7069f","updated":"2021-11-16 15:10:11.000000000","message":"Other than my little comment it looks good to me.\n\nTo make CI pass you\u0027ll want to apply patch [1].\n\nIf this was already tested on a local deployment then we can test it on PS5.\n\n[1] https://review.opendev.org/c/openstack/charm-nova-compute/+/807293/2/test-requirements.txt","commit_id":"2302eaabb5f642e002d87fb111f1e9523c2075c6"},{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"49cd322058727e6197131d39ddce50ced6b4ce3b","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":7,"id":"0ac03ec6_095e3f5d","in_reply_to":"ff1c16bc_7af7069f","updated":"2021-11-17 09:28:08.000000000","message":"Ack","commit_id":"2302eaabb5f642e002d87fb111f1e9523c2075c6"},{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"8b9cafc672010e6cb5d5630cd22c2c834fd3ad8b","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":8,"id":"359904e7_be2aa7bd","updated":"2021-11-18 11:05:23.000000000","message":"\u003e Patch Set 8: Verified-1\n\u003e \n\u003e Build FAILURE by Canonical OpenStack Charm CI.\n\u003e \n\u003e Re-test comment options:\n\u003e \n\u003e - \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n\u003e - \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n\u003e - \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\u003e \n\u003e For more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\u003e \n\u003e \n\u003e - charm-build https://openstack-ci-reports.ubuntu.com/artifacts/d04/817582/8/check/charm-build/d041dba/ : SUCCESS in 8m 57s\n\u003e - osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/a07/817582/8/check/osci-lint/a0712ec/ : SUCCESS in 4m 43s\n\u003e - tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/82d/817582/8/check/tox-py35/82db5fb/ : SUCCESS in 7m 31s\n\u003e - tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/a5a/817582/8/check/tox-py36/a5a1178/ : SUCCESS in 6m 21s\n\u003e - tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/ccb/817582/8/check/tox-py37/ccbf0a6/ : SUCCESS in 6m 29s\n\u003e - tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/4a7/817582/8/check/tox-py38/4a75edd/ : SUCCESS in 7m 25s\n\u003e - vault-impish-xena_rgw vault-impish-xena_rgw : SKIPPED (non-voting)\n\u003e - vault-impish-xena-namespaced vault-impish-xena-namespaced : SKIPPED (non-voting)\n\u003e - vault-hirsute-wallaby_rgw vault-hirsute-wallaby_rgw : SKIPPED\n\u003e - vault-hirsute-wallaby-namespaced vault-hirsute-wallaby-namespaced : SKIPPED\n\u003e - vault-focal-xena_rgw vault-focal-xena_rgw : SKIPPED (non-voting)\n\u003e - vault-focal-xena-namespaced vault-focal-xena-namespaced : SKIPPED (non-voting)\n\u003e - vault-focal-wallaby_rgw vault-focal-wallaby_rgw : SKIPPED\n\u003e - vault-focal-wallaby-namespaced vault-focal-wallaby-namespaced : SKIPPED\n\u003e - vault-focal-victoria_rgw vault-focal-victoria_rgw : SKIPPED\n\u003e - vault-focal-victoria-namespaced vault-focal-victoria-namespaced : SKIPPED\n\u003e - vault-focal-ussuri-ec vault-focal-ussuri-ec : SKIPPED\n\u003e - vault-focal-ussuri_rgw vault-focal-ussuri_rgw : SKIPPED\n\u003e - vault-focal-ussuri-namespaced vault-focal-ussuri-namespaced : SKIPPED\n\u003e - vault-bionic-ussuri https://openstack-ci-reports.ubuntu.com/artifacts/306/817582/8/check/vault-bionic-ussuri/306e295/ : FAILURE in 58m 52s\n\u003e - vault-bionic-ussuri-namespaced vault-bionic-ussuri-namespaced : SKIPPED\n\u003e - vault-bionic-train vault-bionic-train : SKIPPED\n\u003e - vault-bionic-train-namespaced vault-bionic-train-namespaced : SKIPPED\n\u003e - vault-bionic-stein vault-bionic-stein : SKIPPED\n\u003e - vault-bionic-stein-namespaced vault-bionic-stein-namespaced : SKIPPED\n\u003e - vault-bionic-queens vault-bionic-queens : SKIPPED\n\u003e - vault-bionic-queens-namespaced vault-bionic-queens-namespaced : SKIPPED\n\u003e - xenial-mitaka_rgw xenial-mitaka_rgw : SKIPPED\n\u003e - xenial-mitaka-namespaced xenial-mitaka-namespaced : SKIPPED\n\nrecheck","commit_id":"30ee27eb61c4b627d7660aa0fb0f45e36cfc9e92"},{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"0cd0fd0be84e010e3ed3afadea9efcfdf46c7127","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":8,"id":"67d63629_5feb05fa","updated":"2021-11-18 11:06:30.000000000","message":"recheck","commit_id":"30ee27eb61c4b627d7660aa0fb0f45e36cfc9e92"},{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"50e5bb33beeef32b4bcefebe1b181cb3f34df2fa","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":13,"id":"167d3142_082b7902","updated":"2021-12-10 12:09:27.000000000","message":"This PR is set to WIP as it is waiting on https://github.com/juju/charm-helpers/pull/662. \nWhen the PR goes through, I\u0027ll properly pull in the haproxy.cfg change with \"make sync\" and mark the review as active.","commit_id":"72b64fb732044b8b0053c6b0985ae9c552c335b2"},{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"f47ebd1826e9a65eaaf1f9af46a0b57ea381bd00","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":13,"id":"81997eb9_1aaf47c4","updated":"2021-12-10 12:07:58.000000000","message":"This PR is set to WIP as it is waiting on https://github.com/juju/charm-helpers/pull/662. \nWhen the PR goes through, I\u0027ll properly pull in the haproxy.cfg change with \"make sync\" and mark the review as active.","commit_id":"72b64fb732044b8b0053c6b0985ae9c552c335b2"},{"author":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"change_message_id":"541b64a3c97a41f22e732e42a8f1bdfe36bf50bd","unresolved":true,"context_lines":[],"source_content_type":"","patch_set":13,"id":"3f71f14d_3e98e2ac","in_reply_to":"167d3142_082b7902","updated":"2022-01-17 17:53:08.000000000","message":"Hi,the PR was merged on January 10th.","commit_id":"72b64fb732044b8b0053c6b0985ae9c552c335b2"},{"author":{"_account_id":30028,"name":"James Troup","email":"ubuntu@elmo.tasta.io","username":"elmo"},"change_message_id":"5ea20c8695016881a70dec60ad59d4a73f0986cc","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":13,"id":"9ba3b72f_49d0dc72","in_reply_to":"23cf8b87_09a0b26d","updated":"2022-03-02 04:07:37.000000000","message":"Done","commit_id":"72b64fb732044b8b0053c6b0985ae9c552c335b2"},{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"bbf2e0adec47f8b73c77f399fb3fb66af9fd45f5","unresolved":true,"context_lines":[],"source_content_type":"","patch_set":13,"id":"23cf8b87_09a0b26d","in_reply_to":"3f71f14d_3e98e2ac","updated":"2022-01-19 20:56:15.000000000","message":"Thanks Jorge. I have made the review active.","commit_id":"72b64fb732044b8b0053c6b0985ae9c552c335b2"},{"author":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"change_message_id":"eec0785c533a9ba1ae1a670ddd6e647965692819","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":15,"id":"e435d1fb_0353e583","updated":"2022-01-27 12:35:48.000000000","message":"charm-recheck\nDNS errors\n","commit_id":"ba078d487a2697ec70af19007a076bcdcaf4a8c4"},{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"d86acd8913f8ee952ec2907267875c85f3f1b13a","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":16,"id":"70dd902a_c1be533f","updated":"2022-02-07 10:11:57.000000000","message":"Moved the charmhelpers sync changes to a new MR: https://review.opendev.org/c/openstack/charm-ceph-radosgw/+/828099\nMarking as \"workflow: -1\" until we merge it.","commit_id":"78d1457e8b95554b470fb690dc6546da78e1c4fb"},{"author":{"_account_id":30028,"name":"James Troup","email":"ubuntu@elmo.tasta.io","username":"elmo"},"change_message_id":"5ea20c8695016881a70dec60ad59d4a73f0986cc","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":17,"id":"82cf163b_3d787f64","updated":"2022-03-02 04:07:37.000000000","message":"LGTM","commit_id":"31a4584169c1c78e59dc505de4163d528d1cae0a"},{"author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"change_message_id":"ee0e2efee30ea8ee6d4c31d50ffd06a1261f0975","unresolved":false,"context_lines":[],"source_content_type":"","patch_set":17,"id":"a85bb2f0_0d6ee642","updated":"2022-03-04 16:07:51.000000000","message":"LGTM. Thanks for the patch.","commit_id":"31a4584169c1c78e59dc505de4163d528d1cae0a"}],"hooks/ceph_radosgw_context.py":[{"author":{"_account_id":31836,"name":"Facundo Ciccioli","email":"facundo.ciccioli@canonical.com","username":"facu"},"change_message_id":"1d1ee2492eff2f77174d9f4dbb7e7b48b7a67998","unresolved":true,"context_lines":[{"line_number":83,"context_line":""},{"line_number":84,"context_line":"        # for haproxy.conf"},{"line_number":85,"context_line":"        backend_options \u003d {"},{"line_number":86,"context_line":"            \u0027cephradosgw-server\u0027: [{"},{"line_number":87,"context_line":"                \u0027option\u0027: \u0027httpchk GET /swift/healthcheck\u0027,"},{"line_number":88,"context_line":"            }]"},{"line_number":89,"context_line":"        }"}],"source_content_type":"text/x-python","patch_set":7,"id":"4047f35a_c0bd2547","line":86,"updated":"2021-11-16 15:10:11.000000000","message":"I believe it\u0027s time to DRY out this cephradosgw-server magic value. It\u0027s used on line 76 above as well, and from the way it\u0027s used in the template, it\u0027s important that they match.","commit_id":"2302eaabb5f642e002d87fb111f1e9523c2075c6"},{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"49cd322058727e6197131d39ddce50ced6b4ce3b","unresolved":false,"context_lines":[{"line_number":83,"context_line":""},{"line_number":84,"context_line":"        # for haproxy.conf"},{"line_number":85,"context_line":"        backend_options \u003d {"},{"line_number":86,"context_line":"            \u0027cephradosgw-server\u0027: [{"},{"line_number":87,"context_line":"                \u0027option\u0027: \u0027httpchk GET /swift/healthcheck\u0027,"},{"line_number":88,"context_line":"            }]"},{"line_number":89,"context_line":"        }"}],"source_content_type":"text/x-python","patch_set":7,"id":"712f87e5_0905587a","line":86,"in_reply_to":"4047f35a_c0bd2547","updated":"2021-11-17 09:28:08.000000000","message":"I agree.","commit_id":"2302eaabb5f642e002d87fb111f1e9523c2075c6"}],"hooks/charmhelpers/contrib/openstack/templates/haproxy.cfg":[{"author":{"_account_id":27345,"name":"Cornellius Metto","email":"corn.metto@canonical.com","username":"ckmetto"},"change_message_id":"5a268cd72145546b4b04e6c1ed164c62dc162abc","unresolved":true,"context_lines":[{"line_number":72,"context_line":"{% for frontend in frontends -%}"},{"line_number":73,"context_line":"backend {{ service }}_{{ frontend }}"},{"line_number":74,"context_line":"    balance leastconn"},{"line_number":75,"context_line":"    {% if backend_options -%}"},{"line_number":76,"context_line":"    {% if backend_options[service] -%}"},{"line_number":77,"context_line":"    {% for option in backend_options[service] -%}"},{"line_number":78,"context_line":"    {% for key, value in option.items() -%}"}],"source_content_type":"text/x-ttcn-cfg","patch_set":11,"id":"f0a3edd8_885fe0d0","side":"PARENT","line":75,"updated":"2021-12-08 15:31:58.000000000","message":"This file will be reverted. Just showing the changes that need to go to charmhelpers.","commit_id":"3a27c7090e0714d28f1ef0b8e9d22e79573b9445"},{"author":{"_account_id":30028,"name":"James Troup","email":"ubuntu@elmo.tasta.io","username":"elmo"},"change_message_id":"5ea20c8695016881a70dec60ad59d4a73f0986cc","unresolved":false,"context_lines":[{"line_number":72,"context_line":"{% for frontend in frontends -%}"},{"line_number":73,"context_line":"backend {{ service }}_{{ frontend }}"},{"line_number":74,"context_line":"    balance leastconn"},{"line_number":75,"context_line":"    {% if backend_options -%}"},{"line_number":76,"context_line":"    {% if backend_options[service] -%}"},{"line_number":77,"context_line":"    {% for option in backend_options[service] -%}"},{"line_number":78,"context_line":"    {% for key, value in option.items() -%}"}],"source_content_type":"text/x-ttcn-cfg","patch_set":11,"id":"037fa701_a86c1a7a","side":"PARENT","line":75,"in_reply_to":"f0a3edd8_885fe0d0","updated":"2022-03-02 04:07:37.000000000","message":"Done","commit_id":"3a27c7090e0714d28f1ef0b8e9d22e79573b9445"}]}
