)]}'
{"id":"openstack%2Fcharm-cinder-ceph~662986","triplet_id":"openstack%2Fcharm-cinder-ceph~stable%2F19.04~I5dfad42def68634313af05d21ebe61bc229ebc35","project":"openstack/charm-cinder-ceph","branch":"stable/19.04","topic":"bug/1704106","hashtags":[],"change_id":"I5dfad42def68634313af05d21ebe61bc229ebc35","subject":"Add rbd_exclusive_pool support for Ocata and Pike","status":"MERGED","created":"2019-06-04 09:25:39.000000000","updated":"2019-06-18 21:57:26.000000000","submitted":"2019-06-18 21:57:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662986-1560895046647-d473cd44","meta_rev_id":"fc0ac9eef472d8583577dd4eb2997062b4bb28ff","_number":662986,"virtual_id_number":662986,"owner":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"date":"2019-06-18 16:08:07.000000000","_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":1,"date":"2019-06-18 16:52:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":2,"date":"2019-06-18 21:57:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":2,"date":"2019-06-18 21:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-18 18:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":1,"date":"2019-06-18 21:45:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 22:58:32.000000000","updated_by":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"reviewer":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"state":"REVIEWER"},{"updated":"2019-06-18 12:09:19.000000000","updated_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"reviewer":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"state":"REVIEWER"},{"updated":"2019-06-18 16:52:22.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 18:59:14.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REVIEWER"},{"updated":"2019-06-18 21:45:39.000000000","updated_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"reviewer":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"state":"REVIEWER"},{"updated":"2019-06-18 21:57:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa6770106d1c09fc8c421d9656668e98abf0651c","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-04 09:25:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db01e4880cb7d66797f121942fc7e650194f84ab","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-04 09:26:18.000000000","message":"Patch Set 1:\n\ncharm-recheck-full","accounts_in_message":[],"_revision_number":1},{"id":"18274fd0e379ef56c2fbc1a42cb2f4164367d72b","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2019-06-04 09:26:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cfeb5a54bf3326bb3a69e99106f416428b57a3c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 09:49:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/86/662986/1/check/openstack-tox-pep8/0478a97/ : RETRY_LIMIT in 1m 37s\n- openstack-tox-py35 http://logs.openstack.org/86/662986/1/check/openstack-tox-py35/60815fa/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/86/662986/1/check/openstack-tox-cover/3de3861/cover/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"cfbea606510a381bc6172b6fe852b23d47660285","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-06-04 09:56:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/1/10537/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/1/10537/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/1/10537/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/1/10537/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":1},{"id":"b1166ac2fa9f5c1cc86734431bde0a24b5df514b","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-06-04 11:17:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_func_full https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/1/3159/index.html : FAILURE\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/1/3159/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/1/3159/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/1/3159/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":1},{"id":"9c1ea1c55ec5a67e9fb1c2c413d478d63b7071db","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-04 12:15:34.000000000","message":"Patch Set 1:\n\nLooks like the amulet test test_202_cinderceph_cinder_backend_relation is broken on master (it was never updated to support this patch). So im going to submit a patch to master to fix it then add it to this patch.","accounts_in_message":[],"_revision_number":1},{"id":"3c459bccbfbe24ac78967888be036c599c696879","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-04 12:21:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"180c4718551d8d0236c38392efb8f76c131ed3b5","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-04 12:21:33.000000000","message":"Patch Set 2:\n\ncharm-reheck-full","accounts_in_message":[],"_revision_number":2},{"id":"c41af2bd70443465144ce1f691f271ac28be3aad","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-06-04 12:49:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/2/10539/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/2/10539/index.html : FAILURE\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/2/10539/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/2/10539/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":2},{"id":"ce6bd2f59d2c211ab8545300c4d24ba379904bc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 12:53:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/86/662986/2/check/openstack-tox-pep8/8980506/ : SUCCESS in 6m 53s\n- openstack-tox-py35 http://logs.openstack.org/86/662986/2/check/openstack-tox-py35/eb325d1/ : SUCCESS in 5m 00s\n- openstack-tox-cover http://logs.openstack.org/86/662986/2/check/openstack-tox-cover/3b72fb9/cover/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"88eccceee4bc8b37864c6d0f65e31f9e9e93907f","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2019-06-04 13:29:48.000000000","message":"Patch Set 2:\n\nERROR:root:Failed to get logs from application unit: percona-cluster/0","accounts_in_message":[],"_revision_number":2},{"id":"8d57ed1baf7f6b92ee7de86598ebc1917e702dee","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2019-06-04 13:30:18.000000000","message":"Patch Set 2:\n\nthe charm-recheck-full should sort out things","accounts_in_message":[],"_revision_number":2},{"id":"b527915b921d6394c14bfd11f42f126eeb5232d9","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2019-06-06 22:53:51.000000000","message":"Patch Set 2:\n\ncharm-reheck-full","accounts_in_message":[],"_revision_number":2},{"id":"07c054ffc3557c5e93a3d7c200796ee9c4927375","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-07 16:35:53.000000000","message":"Patch Set 2:\n\ncharm-recheck-full","accounts_in_message":[],"_revision_number":2},{"id":"3e3ee399e3a2d6a15c4cf4a89fefc516e5222927","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 17:05:32.000000000","message":"Patch Set 2:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_func_full https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/2/3174/index.html : FAILURE\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/2/3174/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/2/3174/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/2/3174/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":2},{"id":"622697b03416ec2e7827d4e768dcc6c87a2118fc","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2019-06-07 22:58:32.000000000","message":"Patch Set 2:\n\ncharm-recheck-full","accounts_in_message":[],"_revision_number":2},{"id":"a460575c84911df72e5c8ae0962e8218ad5d102e","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:37:21.000000000","message":"Patch Set 2:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_func_full https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/2/3177/index.html : FAILURE\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/2/3177/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/2/3177/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/2/3177/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":2},{"id":"c1f48a49712236695dd39aa3045eef6d38773171","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2019-06-10 14:05:35.000000000","message":"Patch Set 2: Code-Review-1\n\nDEBUG:runner:unexpected relation data in cinder cinder-ceph storage-backend - subordinate_configuration:{\"cinder\": {\"/etc/cinder/cinder.conf\": {\"sections\": {\"cinder-ceph\": [[\"volume_backend_name\", \"cinder-ceph\"], [\"volume_driver\", \"cinder.volume.drivers.rbd.RBDDriver\"], [\"rbd_pool\", \"cinder-ceph\"], [\"rbd_user\", \"cinder-ceph\"], [\"rbd_secret_uuid\", \"c50fe734-13c3-4f2f-9ec1-388625623870\"], [\"rbd_ceph_conf\", \"/var/lib/charm/cinder-ceph/ceph.conf\"], [\"report_discard_supported\", true], [\"rbd_exclusive_cinder_pool\", true]]}}}}\nDEBUG:runner:Exit Code: 1","accounts_in_message":[],"_revision_number":2},{"id":"af292fcb977711aeef1ab461b9fd5f81e38c41fb","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-10 16:55:32.000000000","message":"Patch Set 2:\n\n@freyes that issue was supposed to be addressed by https://review.opendev.org/#/c/663017/ which is synced into this patch but it hasn\u0027t worked so ill dig into why","accounts_in_message":[],"_revision_number":2},{"id":"1476b33a9f58b6a61a134c43ac210ed94e2342a0","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-18 12:07:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4d6ef69c3b9d5d7a4ad781d7d3b734e75f4823c5","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-18 12:08:53.000000000","message":"Patch Set 3:\n\nAssociated fixes for master are in https://review.opendev.org/#/c/663017/","accounts_in_message":[],"_revision_number":3},{"id":"f1c1363ffc36a990f8005eb1cb12fa737531d3c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:27:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/86/662986/3/check/openstack-tox-pep8/396c04c/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/86/662986/3/check/openstack-tox-py35/d2b0848/ : FAILURE in 8m 41s\n- openstack-tox-cover http://logs.openstack.org/86/662986/3/check/openstack-tox-cover/51e9dc4/ : FAILURE in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"33ac0de5e61fb4500777e8325182b0f247cb2ac4","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 12:51:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/3/10644/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/3/10644/index.html : FAILURE\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/3/10644/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/3/10644/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":3},{"id":"717c3e27744ad6951cb98ea988914e4477d0293d","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-18 16:05:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"99c686d051ea8a8a227850c9e59dcae735553865","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2019-06-18 16:08:07.000000000","message":"Patch Set 4:\n\ncharm-recheck-full","accounts_in_message":[],"_revision_number":4},{"id":"75eaca464b05d54ea0c0aa8c2aa76dc1e2dd6a07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:18:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/86/662986/4/check/openstack-tox-pep8/d361e18/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/86/662986/4/check/openstack-tox-py35/66f10fe/ : SUCCESS in 5m 05s\n- openstack-tox-cover http://logs.openstack.org/86/662986/4/check/openstack-tox-cover/df4ae29/cover/ : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4},{"id":"ad68ec13962b400febefcfec20a1fa61bf713aa9","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 16:52:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/4/10646/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/4/10646/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/4/10646/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-cinder-ceph/662986/4/10646/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":4},{"id":"0159d85c7f5e6c0fb3e60f6f29bb4ebe06d70f2a","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 18:56:46.000000000","message":"Patch Set 4:\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_func_full https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/4/3202/index.html : SUCCESS\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/4/3202/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/4/3202/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-cinder-ceph/662986/4/3202/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":4},{"id":"472a3adca05073eab58846e2c3f0576b24cf9d71","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2019-06-18 18:58:48.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM, waiting for the CI","accounts_in_message":[],"_revision_number":4},{"id":"777a3c6edbd35a5c693b53655b0e9704e2c54e79","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2019-06-18 18:59:14.000000000","message":"Patch Set 4: Code-Review+2\n\nah, the CI just returned.","accounts_in_message":[],"_revision_number":4},{"id":"71ae870af043e53b24959523eccb97d63cb7ff6f","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2019-06-18 21:45:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"138f38ee38fb0752fe2ec38e62a2d9266e6af072","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:45:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8e7ccdae4135107620eaf3c56ef0f14d90cfb479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:57:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/86/662986/4/gate/openstack-tox-pep8/78fdd6f/ : SUCCESS in 5m 04s\n- openstack-tox-py35 http://logs.openstack.org/86/662986/4/gate/openstack-tox-py35/08dcbee/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":4},{"id":"bbff3f38723b958dab9ff0d508976919727476d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:57:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2b7ffcaee81045af5893327f0cf2e603d8fd8f5f","revisions":{"3cd8598331e3369dbf3a5b72cd25b2f6cd296ec1":{"kind":"REWORK","_number":1,"created":"2019-06-04 09:25:39.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/86/662986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder-ceph","ref":"refs/changes/86/662986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/1"}}},"commit":{"parents":[{"commit":"1461ee12a5a0964250ad7521f3e20ba7598161f4","subject":"Update stable charms to use stable CH in test reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/1461ee12a5a0964250ad7521f3e20ba7598161f4"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-04-01 16:43:58.000000000","tz":60},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-06-04 09:25:13.000000000","tz":60},"subject":"Add rbd_exclusive_pool support for Ocata and Pike","message":"Add rbd_exclusive_pool support for Ocata and Pike\n\nThe fix has been backported so we can now use it.\n\nChange-Id: I5dfad42def68634313af05d21ebe61bc229ebc35\nCloses-Bug: #1704106\n(cherry picked from commit 616ba364296687bfeb8d9bf35bb0e7628d573e5b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/3cd8598331e3369dbf3a5b72cd25b2f6cd296ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/3cd8598331e3369dbf3a5b72cd25b2f6cd296ec1"}]},"branch":"refs/heads/stable/19.04"},"b82f6b9a2eda95ffab70bd7a181c25902d524943":{"kind":"REWORK","_number":2,"created":"2019-06-04 12:21:15.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/86/662986/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder-ceph","ref":"refs/changes/86/662986/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/2"}}},"commit":{"parents":[{"commit":"1461ee12a5a0964250ad7521f3e20ba7598161f4","subject":"Update stable charms to use stable CH in test reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/1461ee12a5a0964250ad7521f3e20ba7598161f4"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-04-01 16:43:58.000000000","tz":60},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-06-04 12:20:38.000000000","tz":60},"subject":"Add rbd_exclusive_pool support for Ocata and Pike","message":"Add rbd_exclusive_pool support for Ocata and Pike\n\nThe fix has been backported so we can now use it.\n\nAlso includes amulet test fix\n\nChange-Id: I5dfad42def68634313af05d21ebe61bc229ebc35\nCloses-Bug: #1704106\n(cherry picked from commit 616ba364296687bfeb8d9bf35bb0e7628d573e5b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/b82f6b9a2eda95ffab70bd7a181c25902d524943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/b82f6b9a2eda95ffab70bd7a181c25902d524943"}]},"branch":"refs/heads/stable/19.04"},"6e75c73e3c5baff614345c371ee7a025733f3078":{"kind":"REWORK","_number":3,"created":"2019-06-18 12:07:48.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/86/662986/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder-ceph","ref":"refs/changes/86/662986/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/3"}}},"commit":{"parents":[{"commit":"1461ee12a5a0964250ad7521f3e20ba7598161f4","subject":"Update stable charms to use stable CH in test reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/1461ee12a5a0964250ad7521f3e20ba7598161f4"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-04-01 16:43:58.000000000","tz":60},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-06-18 12:06:17.000000000","tz":60},"subject":"Add rbd_exclusive_pool support for Ocata and Pike","message":"Add rbd_exclusive_pool support for Ocata and Pike\n\nThe fix has been backported in Cinder so we\ncan now use it in the charm.\n\nAlso includes amulet test fix and context generator\nfix from commit c3456b7.\n\nChange-Id: I5dfad42def68634313af05d21ebe61bc229ebc35\nCloses-Bug: #1704106\n(cherry picked from commit 616ba364296687bfeb8d9bf35bb0e7628d573e5b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/6e75c73e3c5baff614345c371ee7a025733f3078"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/6e75c73e3c5baff614345c371ee7a025733f3078"}]},"branch":"refs/heads/stable/19.04"},"2b7ffcaee81045af5893327f0cf2e603d8fd8f5f":{"kind":"REWORK","_number":4,"created":"2019-06-18 16:05:36.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/86/662986/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder-ceph","ref":"refs/changes/86/662986/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder-ceph refs/changes/86/662986/4"}}},"commit":{"parents":[{"commit":"1461ee12a5a0964250ad7521f3e20ba7598161f4","subject":"Update stable charms to use stable CH in test reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/1461ee12a5a0964250ad7521f3e20ba7598161f4"}]}],"author":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-04-01 16:43:58.000000000","tz":60},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2019-06-18 16:05:27.000000000","tz":60},"subject":"Add rbd_exclusive_pool support for Ocata and Pike","message":"Add rbd_exclusive_pool support for Ocata and Pike\n\nThe fix has been backported in Cinder so we\ncan now use it in the charm.\n\nAlso includes amulet test fix and context generator\nfix from commit c3456b7.\n\nChange-Id: I5dfad42def68634313af05d21ebe61bc229ebc35\nCloses-Bug: #1704106\n(cherry picked from commit 616ba364296687bfeb8d9bf35bb0e7628d573e5b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/2b7ffcaee81045af5893327f0cf2e603d8fd8f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder-ceph/commit/2b7ffcaee81045af5893327f0cf2e603d8fd8f5f"}]},"branch":"refs/heads/stable/19.04"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
