)]}'
{"id":"openstack%2Fcharm-cinder~854987","triplet_id":"openstack%2Fcharm-cinder~master~I4874777e00fa87884b994b553c1849800af33e62","project":"openstack/charm-cinder","branch":"master","attention_set":{},"removed_from_attention_set":{"8992":{"account":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"last_update":"2022-10-11 11:11:02.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-10-11 11:11:02.000000000","reason":"Change was submitted"},"20870":{"account":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"last_update":"2022-10-11 10:56:06.000000000","reason":"\u003cGERRIT_ACCOUNT_20870\u003e replied on the change","reason_account":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}},"30615":{"account":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"last_update":"2022-10-11 11:11:02.000000000","reason":"Change was submitted"},"5997":{"account":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"last_update":"2022-10-11 11:11:02.000000000","reason":"Change was submitted"},"7198":{"account":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"last_update":"2022-10-11 11:11:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4874777e00fa87884b994b553c1849800af33e62","subject":"Update metadata in setup.cfg","status":"MERGED","created":"2022-08-29 10:48:02.000000000","updated":"2022-10-11 11:11:02.000000000","submitted":"2022-10-11 11:11:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854987","meta_rev_id":"3d3a2e188e7ff1cdb3c58998a1c3c2101bffca66","_number":854987,"virtual_id_number":854987,"owner":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":1,"date":"2022-10-10 12:41:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-11 11:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":2,"date":"2022-10-11 10:56:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":1,"date":"2022-10-11 10:56:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 11:05:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 11:53:30.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-09 11:00:23.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2022-09-09 11:00:23.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-09-09 11:00:23.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":5997,"name":"Walt","display_name":"Hemna","email":"waboring@hemna.com","username":"walter-boring","status":"SAP"},"state":"REVIEWER"},{"updated":"2022-09-13 11:06:16.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"},{"updated":"2022-09-13 11:06:16.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"state":"REVIEWER"}],"messages":[{"id":"c86767e86c2783b8a8dc172cba6aea32969185d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-08-29 10:48:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0294bff841f6e0646ee52e9b6558d3c08651a84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 11:05:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/337c9da30bfb44e091716c38ec7d069f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6992ffdcc1349c78d31f621d9bec9f7 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cdd5fe5be11475b87a46287cf42a9d7 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd55acae6f5145a7b425f7a7e871bd93 : FAILURE in 8m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0731625fcdc403aa1f439d248cf666e : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1341510f7da6243f9f3ca83c91341bba7ceec3b8","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-08-29 11:53:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/edd/854987/1/check/charm-build/edd65ed/ : SUCCESS in 8m 48s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/5ad/854987/1/check/osci-lint/5ada9af/ : SUCCESS in 7m 36s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/396/854987/1/check/tox-py38/396f129/ : SUCCESS in 7m 00s\n- focal-xena https://openstack-ci-reports.ubuntu.com/artifacts/8ae/854987/1/check/focal-xena/8ae1f76/ : SUCCESS in 47m 43s\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/a8e/854987/1/check/jammy-yoga/a8eed03/ : FAILURE in 33m 13s\n- focal-yoga https://openstack-ci-reports.ubuntu.com/artifacts/b80/854987/1/check/focal-yoga/b80c9cb/ : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"530200aa34a8508e915b5734bd5e36ee53c90d07","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-02 08:25:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"495acbd9933fee6f2b71d4c9a3cd7d587f1fa11b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 08:39:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85b961b4140b4eacbee0f2f31f082626\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f5a8fd492647c3a34d4391fcd69f83 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6630425d78e49acab0595e4cf02af6b : SUCCESS in 12m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/947bac6d896740dba1b03e20f8b5b295 : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":1},{"id":"537127998444044a52615bdafd5f86a5b91311c6","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-09-02 11:07:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/862/854987/1/check/charm-build/8625551/ : SUCCESS in 7m 43s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/de8/854987/1/check/osci-lint/de8c700/ : SUCCESS in 8m 06s\n- kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/b5e/854987/1/check/kinetic-zed/b5e92b2/ : FAILURE in 29m 09s (non-voting)\n- jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/e50/854987/1/check/jammy-zed/e50ff5a/ : FAILURE in 30m 27s (non-voting)\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/5de/854987/1/check/jammy-yoga/5de43b3/ : SUCCESS in 1h 19m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8724aa0c923b5795ba2ea6674dbd8a8cb59ee5bb","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-09-13 11:48:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2ae72d5cf26a95b6b4d9ce8a14bacf6839026da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-15 10:25:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"63fd9f4397cab91fe6e53a32a61bf41b8f566a50","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-15 10:25:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e5f437abe39e620fb88edf2b399d9e678f62f1a","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-09-15 10:33:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build charm-build : NODE_FAILURE in 0s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/f27/854987/2/check/osci-lint/f27146b/ : SUCCESS in 7m 09s\n- kinetic-zed kinetic-zed : SKIPPED (non-voting)\n- jammy-zed jammy-zed : SKIPPED (non-voting)\n- jammy-yoga jammy-yoga : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"efead860ced28fefb72f24dce50ab91b04b7281a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 10:37:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b975a09fc7704c7ab57db2ff13ce2da6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92fb62287aa04d3aad56371e89b52f60 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/139d7354cf50454aac8f499022c9fe44 : SUCCESS in 8m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11338a2b92e74d81990d0a015100c272 : POST_FAILURE in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"5e232909f241723a4c105f18e38e1c219f5ba436","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-09-15 11:36:06.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"51fea9f5b1421480f80595966de6a6f176bbe1dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-23 14:34:02.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fd7adf3c18d2383883f44f8e91befc8dd9a54faa","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-23 14:36:30.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"70f76a42805f0aeba4dfb3ca3b434b0e12a7cd9c","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-23 14:36:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"34d1a36ddab9d08abf55cd8fe4af264ddcc9248f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 14:48:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b97dcfd3254b3e8ada78c98865b714\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c312a730224bd3be216dbab6c0bf8c : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/011266fc0f4242eaa523503fa82e7734 : SUCCESS in 9m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8fda0ce2022401bafd04c6f3b170cf0 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"94755598e15fba388a01ce6625c3056ce1b58a6b","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-09-23 15:30:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/6ca/854987/3/check/charm-build/6cad6dc/ : SUCCESS in 8m 28s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/185/854987/3/check/osci-lint/1855f5a/ : SUCCESS in 7m 03s\n- kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/1e3/854987/3/check/kinetic-zed/1e3039c/ : FAILURE in 14m 22s (non-voting)\n- jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/7c4/854987/3/check/jammy-zed/7c46b77/ : FAILURE in 39m 39s (non-voting)\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/844/854987/3/check/jammy-yoga/8449717/ : FAILURE in 46m 01s","accounts_in_message":[],"_revision_number":3},{"id":"170b2bf0fd231249dc5ea03313b8e4d1143709a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-09-26 04:23:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bad47861829bc3af40ceb9ef1c4ce99e4d4a1e5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 04:31:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56016287c8834efabb852d5263c25860\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99eea3348c9c4c57bf612c743c9c99f3 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bea55a7de414d828ea2bf9b766ee780 : SUCCESS in 7m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcb9088a79bc497bb94ce1fd8d600c4d : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":4},{"id":"b5ba322d13637e26df39711ccf46b998df8fbb95","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-09-26 05:13:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/d1f/854987/4/check/charm-build/d1fcd66/ : SUCCESS in 8m 01s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/cf3/854987/4/check/osci-lint/cf36cc6/ : SUCCESS in 7m 24s\n- kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/b6e/854987/4/check/kinetic-zed/b6e9054/ : FAILURE in 22m 19s (non-voting)\n- jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/8b4/854987/4/check/jammy-zed/8b46a63/ : FAILURE in 37m 15s (non-voting)\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/9ab/854987/4/check/jammy-yoga/9ab495f/ : FAILURE in 37m 22s","accounts_in_message":[],"_revision_number":4},{"id":"c7369bfe332a350954b20100c5d08daa472de32e","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-10 11:17:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3ce75c2586eb1029101ed205f7da1d8a5f9df7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 11:31:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7f5ea82cd741fa8054ac2072ea7505\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e5d9b79438441409c43390344bf2805 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/626fe9c81adf4507b41a9c53384980da : SUCCESS in 8m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffdc8c7ac95a408fb6881b05260fd1d4 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":4},{"id":"1606fdbb86ded8e324e287462d2905fad67bac2b","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-10-10 12:41:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/903/854987/4/check/charm-build/9030ee3/ : SUCCESS in 8m 37s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/8b9/854987/4/check/osci-lint/8b98c8b/ : SUCCESS in 7m 31s\n- kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/3ce/854987/4/check/kinetic-zed/3ce0bdb/ : FAILURE in 15m 33s (non-voting)\n- jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/a5f/854987/4/check/jammy-zed/a5f59f3/ : SUCCESS in 59m 26s\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/f74/854987/4/check/jammy-yoga/f748fd2/ : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":4},{"id":"059ad1b103926daaf5f58948d1eedc7e6dec9772","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-10-11 10:04:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eb70118ee52d727e7fde65172a4e93e8ee1c863a","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-10-11 10:56:06.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"23a27dcfa121c6c72761dd64603291dbd0fce393","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 10:56:26.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"caf482c58318c4ee329c0988755ffbf026212867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 11:11:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056fc800f8f44581a3a8d83ff001c4fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e01374c3ce54441a2865c995c9257c7 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f84de8fbee444b3bb6951564f6c8f843 : SUCCESS in 9m 47s","accounts_in_message":[],"_revision_number":4},{"id":"3d3a2e188e7ff1cdb3c58998a1c3c2101bffca66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 11:11:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a4aaa2bef8b631d2dad3165c11fd3e2a5f38039d","revisions":{"4f9028ccc8aa1bd2065d32ac75244a8fd2735d51":{"kind":"REWORK","_number":1,"created":"2022-08-29 10:48:02.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/87/854987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/87/854987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/1"}}},"commit":{"parents":[{"commit":"024de3783fc459fc25596ec6084dd065e945df76","subject":"Add nova section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/024de3783fc459fc25596ec6084dd065e945df76"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-08-29 10:46:55.000000000","tz":330},"committer":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-08-29 10:47:58.000000000","tz":330},"subject":"Update metadata in setup.cfg","message":"Update metadata in setup.cfg\n\nChange-Id: I4874777e00fa87884b994b553c1849800af33e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/4f9028ccc8aa1bd2065d32ac75244a8fd2735d51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/4f9028ccc8aa1bd2065d32ac75244a8fd2735d51"}]},"branch":"refs/heads/master"},"da5156c296e9d3a3b00c2e3d82043aeefc567ff2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-15 10:25:30.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/87/854987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/87/854987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/2"}}},"commit":{"parents":[{"commit":"024de3783fc459fc25596ec6084dd065e945df76","subject":"Add nova section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/024de3783fc459fc25596ec6084dd065e945df76"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-08-29 10:46:55.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2022-09-15 10:25:30.000000000","tz":0},"subject":"Update metadata in setup.cfg","message":"Update metadata in setup.cfg\n\nmaintaining setup.cfg as per other repos[1]\n\n[1] https://review.opendev.org/c/openstack/cinder-tempest-plugin/+/828475\nChange-Id: I4874777e00fa87884b994b553c1849800af33e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/da5156c296e9d3a3b00c2e3d82043aeefc567ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/da5156c296e9d3a3b00c2e3d82043aeefc567ff2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8c96aeae94951437d3efdf69ae12e71a0c9bb63c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-09-23 14:34:02.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/87/854987/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/87/854987/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/3"}}},"commit":{"parents":[{"commit":"024de3783fc459fc25596ec6084dd065e945df76","subject":"Add nova section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/024de3783fc459fc25596ec6084dd065e945df76"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-08-29 10:46:55.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2022-09-23 14:34:01.000000000","tz":0},"subject":"Update metadata in setup.cfg","message":"Update metadata in setup.cfg\n\nwe are using some \"aliases\" that the setuptools docs say \n\"are supported for compatibility reasons\" but their use is not advised[1].\n\n[1] https://setuptools.pypa.io/en/latest/userguide/declarative_config.html#metadata\n\nmaintaining setup.cfg as per other repos[2]\n\n[2] https://review.opendev.org/c/openstack/cinder-tempest-plugin/+/828475\n\nChange-Id: I4874777e00fa87884b994b553c1849800af33e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/8c96aeae94951437d3efdf69ae12e71a0c9bb63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/8c96aeae94951437d3efdf69ae12e71a0c9bb63c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a4aaa2bef8b631d2dad3165c11fd3e2a5f38039d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-09-26 04:23:17.000000000","uploader":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"ref":"refs/changes/87/854987/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/87/854987/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/87/854987/4"}}},"commit":{"parents":[{"commit":"ba21da64f6a547bd44956d71933030d529956bcd","subject":"Re-enable AppArmor enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/ba21da64f6a547bd44956d71933030d529956bcd"}]}],"author":{"name":"tushargite96","email":"tushargite96@gmail.com","date":"2022-08-29 10:46:55.000000000","tz":330},"committer":{"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","date":"2022-09-26 04:23:17.000000000","tz":0},"subject":"Update metadata in setup.cfg","message":"Update metadata in setup.cfg\n\nwe are using some \"aliases\" that the setuptools docs say \n\"are supported for compatibility reasons\" but their use is not advised[1].\n\n[1] https://setuptools.pypa.io/en/latest/userguide/declarative_config.html#metadata\n\nmaintaining setup.cfg as per other repos[2]\n\n[2] https://review.opendev.org/c/openstack/cinder-tempest-plugin/+/828475\n\nChange-Id: I4874777e00fa87884b994b553c1849800af33e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a4aaa2bef8b631d2dad3165c11fd3e2a5f38039d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a4aaa2bef8b631d2dad3165c11fd3e2a5f38039d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"}}]}],"submit_requirements":[]}
