)]}'
{"id":"openstack%2Fcharm-cinder~926643","triplet_id":"openstack%2Fcharm-cinder~stable%2Fyoga~I131e12c53c78957873b6a75a3b1b67b1725b43ed","project":"openstack/charm-cinder","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"8108":{"account":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"last_update":"2024-09-05 10:26:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I131e12c53c78957873b6a75a3b1b67b1725b43ed","subject":"Install lsscsi by default for zed or older","status":"MERGED","created":"2024-08-20 13:38:48.000000000","updated":"2024-09-05 10:26:18.000000000","submitted":"2024-09-05 10:26:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926643","meta_rev_id":"c10a44cb2f3d125940568771db0cafef85f1422c","_number":926643,"virtual_id_number":926643,"owner":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-05 10:26:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":2,"date":"2024-09-05 10:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-05 05:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":1,"date":"2024-09-05 10:10:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-20 13:50:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-05 05:02:11.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"state":"REVIEWER"},{"updated":"2024-09-05 10:10:22.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"}],"messages":[{"id":"50a627d3093345476c027f38a7eda13abaa493d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2024-08-20 13:38:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eced6baa49a448101d3886442eed2afbe260d5c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 13:50:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65e125de3fac41bf806bf687b320e60f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceaa918df19843b89d05060f7849903d : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5056020404e14626a32d74a3f7482c57 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21cd504274f94efa9d867b36937734a7 : FAILURE in 3m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7743b38c5c7a4ba1bf623488c45c444d : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"10e7b05a3ef850c01f7e2debf4b235e37c395544","author":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"date":"2024-09-05 04:44:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b465f25a3a4d5e001fe3eb9db2f9f0b1d9540f88","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2024-09-05 05:02:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9781b7629f964e81e13f243eb3c2cfe592e9f5a7","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2024-09-05 10:10:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e7f12bb4dc9cb823ac07c0ab1ce117b1edccae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 10:10:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9cc52c6ec8ad7a2052a2e21363a940359247707d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 10:26:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea434acf9d6943938b5d34a77091e687\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c61943d45d1d48b5a61e5a12d86cf8cf : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a537997bcf1945afa2fc4b393df68bbc : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c10a44cb2f3d125940568771db0cafef85f1422c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 10:26:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4964b39c507427a9634f82f7b9de90ca00649554","revisions":{"4964b39c507427a9634f82f7b9de90ca00649554":{"kind":"REWORK","_number":1,"created":"2024-08-20 13:38:48.000000000","uploader":{"_account_id":8108,"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","username":"nobuto-m"},"ref":"refs/changes/43/926643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/43/926643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/43/926643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/43/926643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/43/926643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/43/926643/1"}}},"commit":{"parents":[{"commit":"67a7634c4c940ad9711cd307b7c641afaededc4f","subject":"Add keystone audit middleware API logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/67a7634c4c940ad9711cd307b7c641afaededc4f"}]}],"author":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2024-08-20 03:11:46.000000000","tz":540},"committer":{"name":"Nobuto Murata","email":"nobuto.murata@canonical.com","date":"2024-08-20 05:12:23.000000000","tz":540},"subject":"Install lsscsi by default for zed or older","message":"Install lsscsi by default for zed or older\n\nlsscsi is a dependency of os-brick for volume related operations\nincluding creating a boot volume with iSCSI/FC backends. However, it\nwasn\u0027t installed by default until jammy-antelope (lunar) due to a\npackaging bug. It hasn\u0027t been an obvious issue especially when the\ncinder-volume unit and the nova-compute unit are co-located since\nnova-compute charm installs the package. Instead of applying a\nworkaround to all backend charms, let\u0027s apply the workaround in the\ncinder charm since lsscsi is installed by default for later OpenStack\nreleases anyway and it\u0027s a small package.\n\nCloses-Bug: #1939390\nChange-Id: I131e12c53c78957873b6a75a3b1b67b1725b43ed\n(cherry picked from commit 9a6ab28af2f5b01bc5ebc54c7cde3248dd65b53e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/4964b39c507427a9634f82f7b9de90ca00649554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/4964b39c507427a9634f82f7b9de90ca00649554"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
