)]}'
{"id":"openstack%2Fcharm-cinder~990653","triplet_id":"openstack%2Fcharm-cinder~stable%2F2023.2~I09316e2b7b7d17c1ad43ad9ea1894f9524250cf0","project":"openstack/charm-cinder","branch":"stable/2023.2","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2026-06-02 13:10:09.000000000","reason":"\u003cGERRIT_ACCOUNT_6737\u003e replied on the change","reason_account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},"38780":{"account":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"last_update":"2026-06-02 13:19:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09316e2b7b7d17c1ad43ad9ea1894f9524250cf0","subject":"Cleanup Bobcat CI","status":"MERGED","created":"2026-05-29 14:10:47.000000000","updated":"2026-06-02 13:19:33.000000000","submitted":"2026-06-02 13:19:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990653-openstack-charms-ci-cleanup","meta_rev_id":"efb1154ef195723f0d0ffeb14aed7c44bfc9dd0a","_number":990653,"virtual_id_number":990653,"owner":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 13:19:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-02 13:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":2,"date":"2026-06-01 23:33:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-02 13:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 14:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 23:33:10.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-06-02 13:10:09.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"2234aa44375f91e60c73c429738a1d832d9ccb57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-05-29 14:10:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c89ce0d205f744b6b0a4a11ce7eaa166ebbeb932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:21:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95176a79d95f4deba17856d7672cde14\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08990ee3c384cabb4e1fe2156439a2f : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e5a2e2a8486401b82d3757cc83ac8c4 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f6ea362e0394aa887d294d8599e96d4 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0b42d9d87eb96ffbfefb3028a9c48d0862d5317c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-01 16:16:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ff70145f34adf97299c51c0848e7617cc5257daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:27:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a9b75163a574df2b34cb35237ede79b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9a9f9303f74576b04e9159cb65250c : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34f8a7e0b12a41869edc72c4b7a813f1 : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7944800a81a044539f0a5095f76aef4f : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"fd5fadfe388e2d0a4a4ae2a1f5131fb923eb6483","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-06-01 23:33:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df1de49e8530e0e0a3bd54c3ec5e1a34afa47096","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-02 13:10:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"50ada372077a3b0be204e68af9d8daa6d362cd88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 13:10:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2fdabe6ef60932822fbfc97a97f8f2c27c6eb666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 13:19:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5c64f10510d43b5abc7d953b76baeff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c918a933f5b40c5b40ccbd4ac48416e : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d15f1723a41f40e19159305aa487fbf6 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"efb1154ef195723f0d0ffeb14aed7c44bfc9dd0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 13:19:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1adce51dfa2c08cd84b79aeba4329fc34f2e826","revisions":{"494111c3e3e356eb9520d87213d8b28930dbbfaa":{"kind":"REWORK","_number":1,"created":"2026-05-29 14:10:47.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/53/990653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/53/990653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/53/990653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/53/990653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/53/990653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/53/990653/1"}}},"commit":{"parents":[{"commit":"6a295449330e56b8242569f9d6adad54b3848ce1","subject":"Add keystone audit middleware API logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/6a295449330e56b8242569f9d6adad54b3848ce1"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 14:09:55.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 14:09:55.000000000","tz":300},"subject":"Cleanup Bobcat CI","message":"Cleanup Bobcat CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n\nChange-Id: I09316e2b7b7d17c1ad43ad9ea1894f9524250cf0\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/494111c3e3e356eb9520d87213d8b28930dbbfaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/494111c3e3e356eb9520d87213d8b28930dbbfaa"}]},"branch":"refs/heads/stable/2023.2"},"b1adce51dfa2c08cd84b79aeba4329fc34f2e826":{"kind":"REWORK","_number":2,"created":"2026-06-01 16:16:30.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/53/990653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/53/990653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/53/990653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/53/990653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/53/990653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/53/990653/2"}}},"commit":{"parents":[{"commit":"6a295449330e56b8242569f9d6adad54b3848ce1","subject":"Add keystone audit middleware API logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/6a295449330e56b8242569f9d6adad54b3848ce1"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 14:09:55.000000000","tz":300},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-01 16:16:16.000000000","tz":60},"subject":"Cleanup Bobcat CI","message":"Cleanup Bobcat CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n\nChange-Id: I09316e2b7b7d17c1ad43ad9ea1894f9524250cf0\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/b1adce51dfa2c08cd84b79aeba4329fc34f2e826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/b1adce51dfa2c08cd84b79aeba4329fc34f2e826"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
