)]}'
{"id":"openstack%2Fcharm-cinder~990667","triplet_id":"openstack%2Fcharm-cinder~stable%2Fyoga~Id2c2129188e9b36a0cc6d6e57601f362a9a371b7","project":"openstack/charm-cinder","branch":"stable/yoga","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2026-06-02 20:52:28.000000000","reason":"\u003cGERRIT_ACCOUNT_6737\u003e replied on the change","reason_account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},"38780":{"account":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"last_update":"2026-06-02 21:09:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2c2129188e9b36a0cc6d6e57601f362a9a371b7","subject":"Cleanup Yoga CI","status":"MERGED","created":"2026-05-29 15:08:49.000000000","updated":"2026-06-02 21:09:44.000000000","submitted":"2026-06-02 21:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990667-openstack-charms-ci-cleanup","meta_rev_id":"8d399035841a7ab597748106e011aba9dc0e8ad3","_number":990667,"virtual_id_number":990667,"owner":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 21:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-02 20:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":2,"date":"2026-06-01 23:35:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-02 20:52:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 15:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 23:35:10.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-06-02 20:52:28.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"80f3c451f30b4e6bc416d8f768f25f5c4cea2bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-05-29 15:08:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f540c525b83fe2a3b3c16b92abafa390fb0a61c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-05-29 15:16:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5254a84658c571e722662c54eb275fea58a839f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-05-29 15:20:07.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"baa4aa734edec9ebfafa78738bdff23107f3674a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:30:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0580a1f5af946768c84c693321a37cc\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/584e2789439f45a2b5dd7b3a2f6743d0 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54c8e7f9215c497dbfd4df7243f250d8 : SUCCESS in 4m 19s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/d38515da741949a39102ecec903cb540 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6c27c90ccba3e860066e83f4dc123f5e8eebbec8","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-05-29 15:33:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"26f522ff5d10167524d36e2a83b1cb56a712d3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:43:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1345fc416c1418e99d02dd71fa9aef5\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/5eba4d8983fb4e2e9722ecdcf316f42a : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07a5a95719cf455e830afb147aef6f7b : SUCCESS in 3m 55s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/6337ab073cf24bb4b7e15128c35c099d : FAILURE in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"15a9c58b4ce02aac20138cd81c8ac68df2913d45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-05-29 16:50:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1266eadddf21c28c315007cfb32b459a82356719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 17:07:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/642c37383843499cb0605ec78e7bb616\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/44ee7b293b234ef1b97ae64dd98f41d0 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f469ec167f584cb5a49ef534ba827822 : SUCCESS in 4m 04s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/98beee4edd8f45bcb0f68fe57ea1cbfd : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":4},{"id":"9648a3c4e1ec667eafb15b99cf1191a1207bb122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-01 16:17:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"2fca0e161379013f84ec9b20eaaebe1b193e5ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:24:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9c907474a3042f1a58c56634d33488f\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/51209e351c1247848c878d355a2f71de : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87851afd8cba45ddb773f4ac3dafa482 : SUCCESS in 4m 11s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/fd935f97a53d47dc891a0e5bc1b1a186 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":5},{"id":"ba8739e53eaa20aa12d95daf06d8a9512f430d54","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-06-01 23:35:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ab14d70d61d0fae7802602b44960a41d05d29ae3","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-02 20:52:28.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a05985d4c72e103a57d1e78bc7bb808017928564","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 20:53:09.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c11d17ae442d11539ad84b6b1036aa6d9716468d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 21:09:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3a8156fe7841d3806f91764a8072c5\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/754ce58fb07a4b61a5956d5433cd6f0c : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1278efda4e342e8a958ef31c61faa21 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":5},{"id":"8d399035841a7ab597748106e011aba9dc0e8ad3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 21:09:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7813a1a02232f15f866de10d2b5dba152fcfe2a0","revisions":{"9d6694c3f2b0e90cc749a6c08b4329edd92b1be2":{"kind":"REWORK","_number":1,"created":"2026-05-29 15:08:49.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/67/990667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/67/990667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/1"}}},"commit":{"parents":[{"commit":"a297912351892ab2077004677f17650f73e19457","subject":"Make sure sysfsutils is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a297912351892ab2077004677f17650f73e19457"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 15:08:15.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 15:08:15.000000000","tz":300},"subject":"Cleanup Yoga CI","message":"Cleanup Yoga CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - merged-requirements-py38.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n\nChange-Id: Id2c2129188e9b36a0cc6d6e57601f362a9a371b7\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/9d6694c3f2b0e90cc749a6c08b4329edd92b1be2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/9d6694c3f2b0e90cc749a6c08b4329edd92b1be2"}]},"branch":"refs/heads/stable/yoga"},"a29b21fe4de495ab9b047825c3f7b17771c3085c":{"kind":"REWORK","_number":2,"created":"2026-05-29 15:16:30.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/67/990667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/67/990667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/2"}}},"commit":{"parents":[{"commit":"a297912351892ab2077004677f17650f73e19457","subject":"Make sure sysfsutils is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a297912351892ab2077004677f17650f73e19457"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 15:08:15.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 15:16:04.000000000","tz":300},"subject":"Cleanup Yoga CI","message":"Cleanup Yoga CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - merged-requirements-py38.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n* .zuul.yaml:\n   - switch to charm jobs to get py310 support\n\nChange-Id: Id2c2129188e9b36a0cc6d6e57601f362a9a371b7\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a29b21fe4de495ab9b047825c3f7b17771c3085c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a29b21fe4de495ab9b047825c3f7b17771c3085c"}]},"branch":"refs/heads/stable/yoga"},"a370a6d5570fe1af9444a0081b03000e08fb2658":{"kind":"NO_CHANGE","_number":3,"created":"2026-05-29 15:20:07.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/67/990667/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/67/990667/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/3"}}},"commit":{"parents":[{"commit":"a297912351892ab2077004677f17650f73e19457","subject":"Make sure sysfsutils is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a297912351892ab2077004677f17650f73e19457"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 15:08:15.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 15:19:47.000000000","tz":300},"subject":"Cleanup Yoga CI","message":"Cleanup Yoga CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - merged-requirements-py38.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n* .zuul.yaml:\n   - switch to charm jobs to get py310 support\n\nChange-Id: Id2c2129188e9b36a0cc6d6e57601f362a9a371b7\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a370a6d5570fe1af9444a0081b03000e08fb2658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a370a6d5570fe1af9444a0081b03000e08fb2658"}]},"branch":"refs/heads/stable/yoga"},"af517d1ee9037815f24443c15ccafd3c429040be":{"kind":"REWORK","_number":4,"created":"2026-05-29 16:50:54.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/67/990667/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/67/990667/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/4"}}},"commit":{"parents":[{"commit":"a297912351892ab2077004677f17650f73e19457","subject":"Make sure sysfsutils is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a297912351892ab2077004677f17650f73e19457"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 15:08:15.000000000","tz":300},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-05-29 16:50:30.000000000","tz":60},"subject":"Cleanup Yoga CI","message":"Cleanup Yoga CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - merged-requirements-py38.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n* .zuul.yaml:\n   - switch to charm jobs to get py310 support\n* also fixes broken unit test\n\nChange-Id: Id2c2129188e9b36a0cc6d6e57601f362a9a371b7\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/af517d1ee9037815f24443c15ccafd3c429040be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/af517d1ee9037815f24443c15ccafd3c429040be"}]},"branch":"refs/heads/stable/yoga"},"7813a1a02232f15f866de10d2b5dba152fcfe2a0":{"kind":"REWORK","_number":5,"created":"2026-06-01 16:17:36.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/67/990667/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/67/990667/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/67/990667/5"}}},"commit":{"parents":[{"commit":"a297912351892ab2077004677f17650f73e19457","subject":"Make sure sysfsutils is available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/a297912351892ab2077004677f17650f73e19457"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-05-29 15:08:15.000000000","tz":300},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-01 16:17:28.000000000","tz":60},"subject":"Cleanup Yoga CI","message":"Cleanup Yoga CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - merged-requirements-py38.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n* .zuul.yaml:\n   - switch to charm jobs to get py310 support\n* also fixes broken unit test\n\nChange-Id: Id2c2129188e9b36a0cc6d6e57601f362a9a371b7\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/7813a1a02232f15f866de10d2b5dba152fcfe2a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/7813a1a02232f15f866de10d2b5dba152fcfe2a0"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
