)]}'
{"id":"openstack%2Fcharm-cinder~991608","triplet_id":"openstack%2Fcharm-cinder~master~Iadc50350e6cefaf2b9d9b77c68d0e8e70b2960e3","project":"openstack/charm-cinder","branch":"master","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"38780":{"account":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"last_update":"2026-06-05 11:51:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadc50350e6cefaf2b9d9b77c68d0e8e70b2960e3","subject":"Update tox.ini and re-generate requirements","status":"MERGED","created":"2026-06-04 12:37:58.000000000","updated":"2026-06-05 11:51:01.000000000","submitted":"2026-06-05 11:51:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991608-openstack-charms-ci-cleanup","meta_rev_id":"ca04633bb61ac84573dfddf10a744f56b1bdeef8","_number":991608,"virtual_id_number":991608,"owner":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-05 11:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-05 11:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-05 11:41:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 12:51:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-05 11:41:01.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"ef655794d8a1f5338f9c0806c91e513fda64f3fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-04 12:37:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7cce5dc4f4140bc17f23d07a74cf629a161f60c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 12:51:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c337a5b8ed0540c8b36511dda171c82f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d233e35b2b4ce99f90dbd0f0ebba61 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18a4e0abb42a4820ae83608cdd882c00 : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79b3b9b31164446bae90c69066008091 : SUCCESS in 6m 33s\n- charmbuild https://zuul.opendev.org/t/openstack/build/680dc404ca4e41d187c8ef1a016dea3b : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a91ec6426e6553c63becc0edfae3c2de67f88ca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-04 16:05:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"eac53aad7451594f3fc21c9a0c304287ff6aefae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 16:18:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ba53559b094ee0a31ccdb8805ead92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b9d8d0b741e448fad8742395355f8ab : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e2c0d16d98c4ffcb7f7ff45f56e627e : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68949dcb11114ef2932ba3c4e41c64ed : SUCCESS in 6m 32s\n- charmbuild https://zuul.opendev.org/t/openstack/build/008741e437594a0eb596fab36b9341ff : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"e2e4c3cd0a54935ca4e416d046dea995552825f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-04 17:29:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c758e20a8f826d2546f84389094528d665029360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:35:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba5cff95a52045fcabd59e09c55061ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d69213b855f4de5b7218d92d0caf485 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4d654d760c94401960b0d5c3b3a30dd : SUCCESS in 4m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebe9dc7a899341158c0cb458272aefa0 : SUCCESS in 4m 20s\n- charmbuild https://zuul.opendev.org/t/openstack/build/601b813a2e9146f3a461d8b240fd5455 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"4c72568048c80a72a9ab444d49963e9a81337b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-05 09:15:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"ddc2c031e4b2fdfcc0a55e09165bd7f0f8212053","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 09:26:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95468e20fa8f4a2e9c07c6198109da06\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/811e027b65b84cdd9f46d7db03682ba7 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d0436e9f8af489d8e688531d08bc100 : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5278541f7a740bc99c4b65014f1a5f1 : SUCCESS in 5m 12s\n- charmbuild https://zuul.opendev.org/t/openstack/build/5d6c0f29f43e4627862c396947e2ea88 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":4},{"id":"7c4ac102ffdedd6b9196a486faa5efa1748c6cff","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-05 11:41:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"915641204eb9da086fe55adde097d6c8efefe8ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:41:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6778b578ca761b5fc05edf6559c40cb51141a0ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:51:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6483e0ebd1c4c648bb6e78ee3410108\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3346826dd4475a9a1e0847f8026004 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a61ece57b1384b2c968371052c60784a : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":4},{"id":"ca04633bb61ac84573dfddf10a744f56b1bdeef8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:51:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5935b291bac68d4290d1c2889a3f57759bc5e242","revisions":{"c1c98a3b7ff77ecd293c91fd88ece8b71fa2abfd":{"kind":"REWORK","_number":1,"created":"2026-06-04 12:37:58.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/08/991608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/08/991608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/1"}}},"commit":{"parents":[{"commit":"23276db1ccf32b56e85b15e180c10f14289cfcf4","subject":"Add *_proxy envvars to tox passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/23276db1ccf32b56e85b15e180c10f14289cfcf4"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 12:37:39.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 12:37:39.000000000","tz":300},"subject":"Update tox.ini and re-generate requirements","message":"Update tox.ini and re-generate requirements\n\ntox.ini needs to have env for specific Python versions as this\nis needed by the CI Zuul jobs, e.g., py310, py312 etc. Since we\nare using the same CI cleanup procedure as with other charms,\nthis may also modify requirements files, so we re-generate\nthem to be sure they are in sync.\n\nChange-Id: Iadc50350e6cefaf2b9d9b77c68d0e8e70b2960e3\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/c1c98a3b7ff77ecd293c91fd88ece8b71fa2abfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/c1c98a3b7ff77ecd293c91fd88ece8b71fa2abfd"}]},"branch":"refs/heads/master"},"f11af44d582213fc46399a70c545fe082d6a8025":{"kind":"REWORK","_number":2,"created":"2026-06-04 16:05:40.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/08/991608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/08/991608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/2"}}},"commit":{"parents":[{"commit":"23276db1ccf32b56e85b15e180c10f14289cfcf4","subject":"Add *_proxy envvars to tox passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/23276db1ccf32b56e85b15e180c10f14289cfcf4"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 16:05:23.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 16:05:23.000000000","tz":300},"subject":"Update tox.ini and re-generate requirements","message":"Update tox.ini and re-generate requirements\n\ntox.ini needs to have env for specific Python versions as this\nis needed by the CI Zuul jobs, e.g., py310, py312 etc. Since we\nare using the same CI cleanup procedure as with other charms,\nthis may also modify requirements files, so we re-generate\nthem to be sure they are in sync.\n\nChange-Id: Iadc50350e6cefaf2b9d9b77c68d0e8e70b2960e3\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/f11af44d582213fc46399a70c545fe082d6a8025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/f11af44d582213fc46399a70c545fe082d6a8025"}]},"branch":"refs/heads/master"},"0f0caf9f5c9948869bff795c1b33bd15b0b1f677":{"kind":"REWORK","_number":3,"created":"2026-06-04 17:29:52.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/08/991608/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/08/991608/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/3"}}},"commit":{"parents":[{"commit":"23276db1ccf32b56e85b15e180c10f14289cfcf4","subject":"Add *_proxy envvars to tox passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/23276db1ccf32b56e85b15e180c10f14289cfcf4"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 16:05:23.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 17:29:42.000000000","tz":300},"subject":"Update tox.ini and re-generate requirements","message":"Update tox.ini and re-generate requirements\n\ntox.ini needs to have env for specific Python versions as this\nis needed by the CI Zuul jobs, e.g., py310, py312 etc. Since we\nare using the same CI cleanup procedure as with other charms,\nthis may also modify requirements files, so we re-generate\nthem to be sure they are in sync.\n\nChange-Id: Iadc50350e6cefaf2b9d9b77c68d0e8e70b2960e3\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/0f0caf9f5c9948869bff795c1b33bd15b0b1f677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/0f0caf9f5c9948869bff795c1b33bd15b0b1f677"}]},"branch":"refs/heads/master"},"5935b291bac68d4290d1c2889a3f57759bc5e242":{"kind":"REWORK","_number":4,"created":"2026-06-05 09:15:28.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/08/991608/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-cinder","ref":"refs/changes/08/991608/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-cinder refs/changes/08/991608/4"}}},"commit":{"parents":[{"commit":"23276db1ccf32b56e85b15e180c10f14289cfcf4","subject":"Add *_proxy envvars to tox passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/23276db1ccf32b56e85b15e180c10f14289cfcf4"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 16:05:23.000000000","tz":300},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-05 09:15:09.000000000","tz":60},"subject":"Update tox.ini and re-generate requirements","message":"Update tox.ini and re-generate requirements\n\ntox.ini needs to have env for specific Python versions as this\nis needed by the CI Zuul jobs, e.g., py310, py312 etc. Since we\nare using the same CI cleanup procedure as with other charms,\nthis may also modify requirements files, so we re-generate\nthem to be sure they are in sync. Also adds HOME to tox passenv\nas it is needed for zaza/functests.\n\nChange-Id: Iadc50350e6cefaf2b9d9b77c68d0e8e70b2960e3\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/5935b291bac68d4290d1c2889a3f57759bc5e242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-cinder/commit/5935b291bac68d4290d1c2889a3f57759bc5e242"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
