)]}'
{"id":"openstack%2Fcharm-deployment-guide~810515","triplet_id":"openstack%2Fcharm-deployment-guide~master~I07aa35937666aaf000f5c2bc5803bee672c1f126","project":"openstack/charm-deployment-guide","branch":"master","topic":"cdg-ops","hashtags":[],"change_id":"I07aa35937666aaf000f5c2bc5803bee672c1f126","subject":"Add operation: Remove a cloud database node","status":"ABANDONED","created":"2021-09-22 20:17:45.000000000","updated":"2023-05-31 09:42:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"ec1ea6537829e6e045281c4a84d35b82ff9b60e0","_number":810515,"virtual_id_number":810515,"owner":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-31 09:31:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2021-09-22 20:25:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 11:42:12.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"CC"}],"messages":[{"id":"3fe2df67c637f91c9d7e9cd35abc8a6121b14805","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2021-09-22 20:17:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94f78a0b3b3e538291d67dace5e560821c8ce74d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 20:25:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/fc5a81ed6cd94fec83d0972a8ab3949e : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"589adb8386d36780c3da2e654600b59490b898a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2021-09-22 22:08:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ee72eecc0912ce2ebef88930c6fb9a2c5b6b96f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 22:13:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/3e4b70af20aa4ab686f6b3b6a1ff7e16 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8343aa62335507abfdae2809f804275e7c6515f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2021-09-29 17:53:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1172671a1ed4b851c76b0cdc3d3198d5694303aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 17:59:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/61bb01e2761841bd856a7b56a5ac3a96 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":3},{"id":"c06ab84fc4d851bbed975d22e447d4c9dc493ea3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2022-01-04 11:42:12.000000000","message":"Abandoned\n\nThis review is \u003e 12 weeks without comment, and failed testing the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3},{"id":"1b2727c283e4a284d922c1df2a3ab82a39d133d4","tag":"autogenerated:gerrit:restore","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2023-05-31 09:30:58.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"280ee94208a364da5682fb5a363a61d0548fb2f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 09:31:18.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/charm-deployment-guide for 810515,3","accounts_in_message":[],"_revision_number":3},{"id":"ec1ea6537829e6e045281c4a84d35b82ff9b60e0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2023-05-31 09:42:12.000000000","message":"Abandoned\n\nRe-abandoning as ops doc items have moved to the charm guide.  This review is now at https://review.opendev.org/c/openstack/charm-guide/+/884853","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5c791ed734f5edc44345f0e0ae0d7b500e58d372","revisions":{"28448ef408d538d0b65410993ec7b7df2892351a":{"kind":"REWORK","_number":1,"created":"2021-09-22 20:17:45.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/15/810515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-deployment-guide","ref":"refs/changes/15/810515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/1"}}},"commit":{"parents":[{"commit":"8197f0959792c61e1f300533c5c5a9112bfa2552","subject":"Merge \"Clarify Neutron AZ\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/8197f0959792c61e1f300533c5c5a9112bfa2552"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2021-09-22 20:17:28.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2021-09-22 20:17:28.000000000","tz":-240},"subject":"Add operation: Remove a cloud database node","message":"Add operation: Remove a cloud database node\n\nChange-Id: I07aa35937666aaf000f5c2bc5803bee672c1f126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/28448ef408d538d0b65410993ec7b7df2892351a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/28448ef408d538d0b65410993ec7b7df2892351a"}]},"branch":"refs/heads/master"},"381d9a5cbdc0e82d8d9741252dd025c82b8439d0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-22 22:08:22.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/15/810515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-deployment-guide","ref":"refs/changes/15/810515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/2"}}},"commit":{"parents":[{"commit":"8197f0959792c61e1f300533c5c5a9112bfa2552","subject":"Merge \"Clarify Neutron AZ\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/8197f0959792c61e1f300533c5c5a9112bfa2552"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2021-09-22 20:17:28.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2021-09-22 22:07:11.000000000","tz":-240},"subject":"Add operation: Remove a cloud database node","message":"Add operation: Remove a cloud database node\n\nAdd a cloud operation to the guide.\n\nCloses-Bug: #1889792\nChange-Id: I07aa35937666aaf000f5c2bc5803bee672c1f126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/381d9a5cbdc0e82d8d9741252dd025c82b8439d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/381d9a5cbdc0e82d8d9741252dd025c82b8439d0"}]},"branch":"refs/heads/master"},"5c791ed734f5edc44345f0e0ae0d7b500e58d372":{"kind":"REWORK","_number":3,"created":"2021-09-29 17:53:17.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/15/810515/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-deployment-guide","ref":"refs/changes/15/810515/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-deployment-guide refs/changes/15/810515/3"}}},"commit":{"parents":[{"commit":"8197f0959792c61e1f300533c5c5a9112bfa2552","subject":"Merge \"Clarify Neutron AZ\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/8197f0959792c61e1f300533c5c5a9112bfa2552"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2021-09-22 20:17:28.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2021-09-29 17:52:58.000000000","tz":-240},"subject":"Add operation: Remove a cloud database node","message":"Add operation: Remove a cloud database node\n\nAdd a cloud operation to the guide.\n\nCloses-Bug: #1889792\nChange-Id: I07aa35937666aaf000f5c2bc5803bee672c1f126\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/5c791ed734f5edc44345f0e0ae0d7b500e58d372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-deployment-guide/commit/5c791ed734f5edc44345f0e0ae0d7b500e58d372"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
