)]}'
{"id":"openstack%2Fcharm-designate-bind~746064","triplet_id":"openstack%2Fcharm-designate-bind~stable%2F20.08~Iff2815c3dd5c82f1a2e6d0cdebb6eac801c157ff","project":"openstack/charm-designate-bind","branch":"stable/20.08","topic":"release-20.08","hashtags":[],"change_id":"Iff2815c3dd5c82f1a2e6d0cdebb6eac801c157ff","subject":"Updates for stable branch creation","status":"MERGED","created":"2020-08-13 10:31:16.000000000","updated":"2020-08-13 12:28:23.000000000","submitted":"2020-08-13 12:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746064-1597321703064-39e11680","meta_rev_id":"ca7f8e46758ff50c2732e58a38a5b0f88a4ac825","_number":746064,"virtual_id_number":746064,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-13 12:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 11:49:56.000000000","_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-13 11:49:56.000000000","_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 11:49:56.000000000","updated_by":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"reviewer":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"state":"REVIEWER"},{"updated":"2020-08-13 12:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"24b506d3ea6b478146e0c119a0a9ff47d2cf180b","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2020-08-13 10:31:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e14dfbdb016324c56dbc3527e87d19fcd16751d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 10:41:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb12d9c19c3e44dfb09ed794a0aa7607 : SUCCESS in 2m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c833189934c5467a8fb150bafc485dfb : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ab47526d2824897b6edbe382979abc5 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23b8ed3a0ccf47feb2cc054adc465469 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0b765b47561459296806d5eda2d89b3 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1392525659e41a39f22cba0fc2bd077feb79ba32","author":{"_account_id":22799,"name":"Andrew McLeod","email":"andrew.mcleod@canonical.com","username":"admcleod"},"date":"2020-08-13 11:49:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"ff85d25f6f135fc19388f985a84f32bb73216507","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 11:50:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fc1d9727f882b0eacd72da7693ddc4fbc7f4112","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:28:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ceec402d9e54c10846b67cfd05bf4a4 : SUCCESS in 3m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f60d89d8c5974b398333a5451eebdf1c : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/676b9cbac6eb465fbeaf1a1247ef92dc : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f87d8245c7924eb89ce02d211652a4da : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2dd10a3d4b7059d12be6060afdd922e208171999","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 12:28:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2351a757ff565226bfbcf0ca65c748dde5d951d7","revisions":{"2351a757ff565226bfbcf0ca65c748dde5d951d7":{"kind":"REWORK","_number":1,"created":"2020-08-13 10:31:16.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/64/746064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-designate-bind","ref":"refs/changes/64/746064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/64/746064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/64/746064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/64/746064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-designate-bind refs/changes/64/746064/1"}}},"commit":{"parents":[{"commit":"c72f509c0c6da8a59d38e9a33b5748344762a180","subject":"Release sync for 20.08","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/c72f509c0c6da8a59d38e9a33b5748344762a180"}]}],"author":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:31:04.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2020-08-13 10:31:14.000000000","tz":60},"subject":"Updates for stable branch creation","message":"Updates for stable branch creation\n\nSet default branch for git review/gerrit.\n\nSwitch tests to stable.\n\nSwitch to using stable charm-helpers branch.\n\nSwitch to using stable charm.openstack branch.\n\nChange-Id: Iff2815c3dd5c82f1a2e6d0cdebb6eac801c157ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/2351a757ff565226bfbcf0ca65c748dde5d951d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/2351a757ff565226bfbcf0ca65c748dde5d951d7"}]},"branch":"refs/heads/stable/20.08"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
