)]}'
{"id":"openstack%2Fcharm-designate-bind~861417","triplet_id":"openstack%2Fcharm-designate-bind~stable%2Fzed~I430d61973ce93a81ed9430295f00cd27208dacfa","project":"openstack/charm-designate-bind","branch":"stable/zed","topic":"zed-stable-updates","attention_set":{},"removed_from_attention_set":{"8992":{"account":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"last_update":"2022-10-27 01:57:34.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"20870":{"account":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"last_update":"2022-10-27 12:17:01.000000000","reason":"Change was submitted"},"2424":{"account":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"last_update":"2022-10-27 12:17:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I430d61973ce93a81ed9430295f00cd27208dacfa","subject":"Updates for zed stable branch creation","status":"MERGED","created":"2022-10-14 16:04:55.000000000","updated":"2022-10-27 12:17:01.000000000","submitted":"2022-10-27 12:17:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861417-zed-stable-updates","meta_rev_id":"8d6fc4a7052fe8d947cdb92cf4b82add1cc498d6","_number":861417,"virtual_id_number":861417,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-27 12:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"all":[{"value":2,"date":"2022-10-27 12:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"all":[{"value":1,"date":"2022-10-27 12:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-14 16:19:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-14 17:54:54.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-26 16:08:21.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"CC"},{"updated":"2022-10-27 12:09:51.000000000","updated_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"reviewer":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"state":"REVIEWER"}],"messages":[{"id":"0ebc9a0970511dad364317c57e410eea9cc0b0f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-10-14 16:04:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3483cabefe486861b972f047bda612a9245ce293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 16:19:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/497f399e20a4422dbec313556bd237aa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299f1446fa26448ab003b31cfd96159d : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/419aa5bcc9d1405f84eadb665ef8b9c9 : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab908de931274533a859738dabbb8dc6 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"491ca5cf3635c3e0dd7a9e623ed74edc3058cc2f","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-10-14 17:54:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/45f/861417/1/check/charm-build/45fbf6c/ : SUCCESS in 11m 39s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/677/861417/1/check/osci-lint/677aa3b/ : SUCCESS in 7m 40s\n- kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/c22/861417/1/check/kinetic-zed/c22280b/ : FAILURE in 14m 27s (non-voting)\n- jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/2ac/861417/1/check/jammy-zed/2ac4d49/ : FAILURE in 13m 10s\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/f67/861417/1/check/jammy-yoga/f67e752/ : FAILURE in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4c27d9521b874f2d0e9d3dc36611be37f5d3e985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2022-10-21 02:31:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e95c450bb92edc6b329af8dbe6782b884607b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 02:46:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e097dad12a334265ab345e791ffb9027\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55e72f6cc8d5450e8fb1097cb372bc4c : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f11b411b97eb464b82f4cffc94746197 : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c9f4f3e7de648e98c55be1ac4be3556 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"e4eea41749cb5b1c0fba04b00f248ab6e89d9e5a","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-10-21 07:20:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/699/861417/2/check/charm-build/699b360/ : SUCCESS in 12m 20s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/57d/861417/2/check/osci-lint/57da04d/ : SUCCESS in 7m 12s\n- kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/649/861417/2/check/kinetic-zed/649ef64/ : FAILURE in 16m 13s (non-voting)\n- jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/8bd/861417/2/check/jammy-zed/8bd4444/ : SUCCESS in 43m 35s\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/67f/861417/2/check/jammy-yoga/67f54c2/ : FAILURE in 43m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0d40866c4eb39c34b40a65e6dc6daa2aa3b4ef4c","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-10-26 16:08:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c33bbc39111d50e73c936f9a94067261d7a9945f","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-10-26 16:19:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f1031ebea6a579db7f4135c83d9cb22c78eb467","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-10-26 16:29:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e3e2d2fa18766a03c9e246fba42bc56c20b0335","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-10-26 22:36:52.000000000","message":"Patch Set 2:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/bc3/861417/2/check/charm-build/bc38ac7/ : SUCCESS in 14m 24s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/71f/861417/2/check/osci-lint/71f6678/ : SUCCESS in 11m 05s\n- kinetic-zed https://openstack-ci-reports.ubuntu.com/artifacts/139/861417/2/check/kinetic-zed/13900bc/ : FAILURE in 15m 53s (non-voting)\n- jammy-zed https://openstack-ci-reports.ubuntu.com/artifacts/682/861417/2/check/jammy-zed/682c9b5/ : FAILURE in 44m 33s\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/04b/861417/2/check/jammy-yoga/04bf55b/ : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":2},{"id":"901919d309b825459176fff50454711cbeb0f0c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2022-10-27 01:44:14.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"36534e34e5e6e8e9223667df5a412b9d3b2ffbb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 01:57:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c51b001b34441688bd6a8104d8eff8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78b41a4b30474e86a74eeaf264ebdd01 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24eb5c2b869142219343762e31058681 : SUCCESS in 5m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3258039399dd49fb9cf461467e94b2e7 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"eea7f1d0fa881b5872dd70997737c1032f8c1e6c","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2022-10-27 12:09:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7f23b62548ab30c2e9193ff7810ac1931435c11f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:10:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"641656faf38e5546a8114567d1ecee0c6373a549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:17:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d946ec0b2b2d42ef99349e80b8e5de63\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5380f77f2343078e9fe10322ce4d3b : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7424b4ad00614a8bbf320ffb8fc57c31 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":3},{"id":"8d6fc4a7052fe8d947cdb92cf4b82add1cc498d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:17:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"52c9a83e490d13e048e86d5abcb34aed19a78542","revisions":{"f35a412d6e2144cf3f1f29d5e37947116ecfe609":{"kind":"REWORK","_number":1,"created":"2022-10-14 16:04:55.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/17/861417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-designate-bind","ref":"refs/changes/17/861417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/1"}}},"commit":{"parents":[{"commit":"ca2f0a404c44fcbe8758fb491d23ff04a5ff7758","subject":"Fix s390x and pp64el builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/ca2f0a404c44fcbe8758fb491d23ff04a5ff7758"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2022-10-14 16:04:50.000000000","tz":60},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2022-10-14 16:04:54.000000000","tz":60},"subject":"Updates for zed stable branch creation","message":"Updates for zed stable branch creation\n\n- Set default branch for git review/gerrit to stable/zed\n- Switch tests to stable.\n- Switch to using stable charm-helpers branch.\n- Switch to using stable charm.openstack branch.\n- Switch to using stable zaza, zaza-openstack-tests\n  branch\n- (reactive charms) Add build.lock file\n- (classic charms) make sync\n- (reactive: not reactive plugin): lock charm-tools \u003c 3.1\n- (reactive: with reactive plugin): lock charm snap to 3.x/stable\n\nChange-Id: I430d61973ce93a81ed9430295f00cd27208dacfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/f35a412d6e2144cf3f1f29d5e37947116ecfe609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/f35a412d6e2144cf3f1f29d5e37947116ecfe609"}]},"branch":"refs/heads/stable/zed"},"f49a46def530bfcbc568a63cd46a9cbda1d642be":{"kind":"REWORK","_number":2,"created":"2022-10-21 02:31:48.000000000","uploader":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"ref":"refs/changes/17/861417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-designate-bind","ref":"refs/changes/17/861417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/2"}}},"commit":{"parents":[{"commit":"ca2f0a404c44fcbe8758fb491d23ff04a5ff7758","subject":"Fix s390x and pp64el builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/ca2f0a404c44fcbe8758fb491d23ff04a5ff7758"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2022-10-14 16:04:50.000000000","tz":60},"committer":{"name":"Billy Olsen","email":"billy.olsen@gmail.com","date":"2022-10-21 02:31:46.000000000","tz":-420},"subject":"Updates for zed stable branch creation","message":"Updates for zed stable branch creation\n\n- Set default branch for git review/gerrit to stable/zed\n- Switch tests to stable.\n- Switch to using stable charm-helpers branch.\n- Switch to using stable charm.openstack branch.\n- Switch to using stable zaza, zaza-openstack-tests\n  branch\n- (reactive charms) Add build.lock file\n- (classic charms) make sync\n- (reactive: not reactive plugin): lock charm-tools \u003c 3.1\n- (reactive: with reactive plugin): lock charm snap to 3.x/stable\n\nChange-Id: I430d61973ce93a81ed9430295f00cd27208dacfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/f49a46def530bfcbc568a63cd46a9cbda1d642be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/f49a46def530bfcbc568a63cd46a9cbda1d642be"}]},"branch":"refs/heads/stable/zed"},"52c9a83e490d13e048e86d5abcb34aed19a78542":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-27 01:44:14.000000000","uploader":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"ref":"refs/changes/17/861417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-designate-bind","ref":"refs/changes/17/861417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-designate-bind refs/changes/17/861417/3"}}},"commit":{"parents":[{"commit":"ca2f0a404c44fcbe8758fb491d23ff04a5ff7758","subject":"Fix s390x and pp64el builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/ca2f0a404c44fcbe8758fb491d23ff04a5ff7758"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2022-10-14 16:04:50.000000000","tz":60},"committer":{"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","date":"2022-10-27 01:43:49.000000000","tz":-180},"subject":"Updates for zed stable branch creation","message":"Updates for zed stable branch creation\n\n- Set default branch for git review/gerrit to stable/zed\n- Switch tests to stable.\n- Switch to using stable charm-helpers branch.\n- Switch to using stable charm.openstack branch.\n- Switch to using stable zaza, zaza-openstack-tests\n  branch\n- (reactive charms) Add build.lock file\n- (classic charms) make sync\n- (reactive: not reactive plugin): lock charm-tools \u003c 3.1\n- (reactive: with reactive plugin): lock charm snap to 3.x/stable\n\nFunc-Test-Pr: https://github.com/openstack-charmers/zaza-openstack-tests/pull/971\nChange-Id: I430d61973ce93a81ed9430295f00cd27208dacfa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/52c9a83e490d13e048e86d5abcb34aed19a78542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/52c9a83e490d13e048e86d5abcb34aed19a78542"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
