)]}'
{"id":"openstack%2Fcharm-designate-bind~990903","triplet_id":"openstack%2Fcharm-designate-bind~stable%2F2023.1~Ifb5fafca200e8feadbb0a79e3b855e52c32eede5","project":"openstack/charm-designate-bind","branch":"stable/2023.1","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2026-06-02 16:17:55.000000000","reason":"\u003cGERRIT_ACCOUNT_6737\u003e replied on the change","reason_account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},"38780":{"account":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"last_update":"2026-06-02 16:26:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb5fafca200e8feadbb0a79e3b855e52c32eede5","subject":"Cleanup Antelope CI","status":"MERGED","created":"2026-06-01 10:55:06.000000000","updated":"2026-06-02 16:26:59.000000000","submitted":"2026-06-02 16:26:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990903-openstack-charms-ci-cleanup","meta_rev_id":"7c16645d6795c218a041cb2332b6c3fb93b885da","_number":990903,"virtual_id_number":990903,"owner":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 16:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-02 16:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":2,"date":"2026-06-01 23:27:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-02 16:17:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-01 11:06:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 23:27:41.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-06-02 16:17:55.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"9aa690bf801ee7b3f38e259d297a836d76976c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-01 10:55:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e91c58b1d9e47f76b1d5d343f3a7f8fb2d02aec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 11:06:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4a3df605d9b4878b9fe6e5f0d8e4491\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d673d7a99fa84bf9b8a7ae919ff56cbe : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fd26201d8444a1f88b3a65fc7d11f3d : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/836db3c971174fd7824c98d3e0cd0665 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"32df44b36df99764a03f73ca4c9312ca18746a4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-01 16:26:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"be583202da2a2efa4933eb03556812132850e3a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:40:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f70aaaa90c134516be17d9166922bf2c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6693dde3f2949e489412e68302626a6 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ade9335d78d4fc9bc989becd0fe9083 : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53091ed6e50044e1b074b17f5b2b5e3a : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":2},{"id":"282c9b502a75ab5b9a13c4cca4abe03c572bb3c1","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-06-01 23:27:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b40bd603a890cdfe7ed2cf43a10258ebd28f415b","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-02 16:17:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d15dfdecd83e316850f02f26c966284719513df8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:18:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25bbe32a989cf6e23340ac7b315680ebc9f2beb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:26:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fb1e58e899a4eb481ba6b8ceb1ae9ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adcc19ea998c43a0b14c7c4aa8265391 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8ff8d0bb29d4427807d7faf4fa99b69 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"7c16645d6795c218a041cb2332b6c3fb93b885da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:26:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c91c2e3a81e57e8ac17f86ca299fd175c320091d","revisions":{"112fae9eb0a239b2585ed00d28af8cd7cd440a0d":{"kind":"REWORK","_number":1,"created":"2026-06-01 10:55:06.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/03/990903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-designate-bind","ref":"refs/changes/03/990903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/03/990903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/03/990903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/03/990903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-designate-bind refs/changes/03/990903/1"}}},"commit":{"parents":[{"commit":"8fed2f9eaf28cbb6ed59296f4abed9887d8a80a4","subject":"Bump up Charm-helpers revision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/8fed2f9eaf28cbb6ed59296f4abed9887d8a80a4"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 10:54:50.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 10:54:50.000000000","tz":300},"subject":"Cleanup Antelope CI","message":"Cleanup Antelope CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini and src/tox.ini\n\nChange-Id: Ifb5fafca200e8feadbb0a79e3b855e52c32eede5\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/112fae9eb0a239b2585ed00d28af8cd7cd440a0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/112fae9eb0a239b2585ed00d28af8cd7cd440a0d"}]},"branch":"refs/heads/stable/2023.1"},"c91c2e3a81e57e8ac17f86ca299fd175c320091d":{"kind":"REWORK","_number":2,"created":"2026-06-01 16:26:11.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/03/990903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-designate-bind","ref":"refs/changes/03/990903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/03/990903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/03/990903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-designate-bind refs/changes/03/990903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-designate-bind refs/changes/03/990903/2"}}},"commit":{"parents":[{"commit":"8fed2f9eaf28cbb6ed59296f4abed9887d8a80a4","subject":"Bump up Charm-helpers revision","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/8fed2f9eaf28cbb6ed59296f4abed9887d8a80a4"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 10:54:50.000000000","tz":300},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-01 16:26:03.000000000","tz":60},"subject":"Cleanup Antelope CI","message":"Cleanup Antelope CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini and src/tox.ini\n\nChange-Id: Ifb5fafca200e8feadbb0a79e3b855e52c32eede5\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/c91c2e3a81e57e8ac17f86ca299fd175c320091d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate-bind/commit/c91c2e3a81e57e8ac17f86ca299fd175c320091d"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
