)]}'
{"id":"openstack%2Fcharm-designate~848211","triplet_id":"openstack%2Fcharm-designate~master~Ib9843bda9c6967ce50cec8d721079bd70f7f3040","project":"openstack/charm-designate","branch":"master","topic":"bug/1925233","attention_set":{},"removed_from_attention_set":{"12549":{"account":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"last_update":"2022-07-20 08:15:31.000000000","reason":"\u003cGERRIT_ACCOUNT_12549\u003e replied on the change","reason_account":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"}},"34959":{"account":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"last_update":"2022-07-20 08:25:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9843bda9c6967ce50cec8d721079bd70f7f3040","subject":"Disable designate-mdns service until the configuration has been rendered","status":"MERGED","created":"2022-06-30 03:41:07.000000000","updated":"2022-07-20 08:25:33.000000000","submitted":"2022-07-20 08:25:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848211-bug/1925233","meta_rev_id":"6e38e80ba609d18e810f7cae3011762ebe9e38d5","_number":848211,"virtual_id_number":848211,"owner":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":1,"date":"2022-07-19 12:47:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-20 08:25:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"all":[{"value":2,"date":"2022-07-20 08:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"all":[{"value":1,"date":"2022-07-20 08:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-30 03:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 04:56:52.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-01 02:27:53.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"CC"},{"updated":"2022-07-18 08:52:40.000000000","updated_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"reviewer":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"state":"REVIEWER"}],"messages":[{"id":"fc5d229a15ef82c5478f98ea779d9a866688de0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"date":"2022-06-30 03:41:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd6f37b995eb5b8e991b260b34d4b290ae1de03e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 03:49:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62603984fdb2451fa837ca10be25e486 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd5052643a364dc2bd7ea44db687d939 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/859fd7e776644130bef455c7b9999aec : FAILURE in 2m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bbaa2a8d62a4d6a84e5eb696044f8cb : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e16ceab64be8bb56c674566009a64a368416d67e","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-06-30 04:56:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/ba9/848211/1/check/charm-build/ba975f5/ : SUCCESS in 17m 23s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/f30/848211/1/check/osci-lint/f30f9af/ : SUCCESS in 9m 42s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/8f3/848211/1/check/tox-py38/8f32610/ : SUCCESS in 10m 22s\n- focal-xena https://openstack-ci-reports.ubuntu.com/artifacts/e65/848211/1/check/focal-xena/e656559/ : SUCCESS in 50m 57s\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/f0a/848211/1/check/jammy-yoga/f0a8ffb/ : POST_FAILURE in 29m 05s (non-voting)\n- focal-yoga https://openstack-ci-reports.ubuntu.com/artifacts/6b3/848211/1/check/focal-yoga/6b33385/ : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3f3110055cf23db9f6d767a2d9b6d3bcff8d8d51","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2022-07-18 08:52:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"accab1fe8acbf4e83028688c2529d93790744b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"date":"2022-07-18 13:36:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2a1d5a11c792a501c6c268bb05d31e96c9eb892","author":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"date":"2022-07-18 13:37:08.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ec462f568e2b595f1c40079bae6b3d5c9a44f3c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 13:44:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dd23f7071cb406fb5d75bd1bc7eca4e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1e8af15b44b46c983f029249e410426 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b71bc6d6ceb4999a26d673a276bfe83 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18790e0a87554eaea880e8463bf75347 : FAILURE in 5m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95ff9078c82d46c69b1672eda46d7ae9 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":2},{"id":"0131d959e6be2a42beb83ad490178769f58b6c40","author":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"date":"2022-07-18 20:46:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3c1a80e07d2c3dad234c388cc8a437c86710c90","author":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"date":"2022-07-19 06:21:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"906d941a6bb3805d6db3b3a1ebc8a8229c296c66","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-19 09:09:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/c57/848211/2/check/charm-build/c570515/ : SUCCESS in 16m 51s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/c85/848211/2/check/osci-lint/c852771/ : SUCCESS in 5m 43s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/a02/848211/2/check/tox-py38/a027089/ : SUCCESS in 6m 59s\n- focal-xena https://openstack-ci-reports.ubuntu.com/artifacts/3d0/848211/2/check/focal-xena/3d00688/ : SUCCESS in 43m 00s\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/c9c/848211/2/check/jammy-yoga/c9c4ba4/ : FAILURE in 38m 49s (non-voting)\n- focal-yoga https://openstack-ci-reports.ubuntu.com/artifacts/4ef/848211/2/check/focal-yoga/4ef3b72/ : FAILURE in 2h 28m 15s","accounts_in_message":[],"_revision_number":2},{"id":"632c7332c68f9211799900a909d72580031d3c2e","author":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"date":"2022-07-19 09:25:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10ff44e3daefde8950b95ac4a72fa4a067d83782","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2022-07-19 12:47:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/19d/848211/2/check/charm-build/19d0802/ : SUCCESS in 18m 14s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/520/848211/2/check/osci-lint/520475b/ : SUCCESS in 7m 01s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/d23/848211/2/check/tox-py38/d23683e/ : SUCCESS in 7m 18s\n- focal-xena https://openstack-ci-reports.ubuntu.com/artifacts/dde/848211/2/check/focal-xena/ddeeb98/ : SUCCESS in 48m 06s\n- jammy-yoga https://openstack-ci-reports.ubuntu.com/artifacts/c10/848211/2/check/jammy-yoga/c10312c/ : FAILURE in 40m 16s (non-voting)\n- focal-yoga https://openstack-ci-reports.ubuntu.com/artifacts/b66/848211/2/check/focal-yoga/b66c38b/ : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":2},{"id":"378b4117323546ae3a7fa25a51af64a0b91ebaed","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2022-07-20 08:15:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae1d82b842517ed159465ccc6dfa998bed2d192d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 08:15:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"12978e0ff13e247be02e27b77237d7d0bf4e5689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 08:25:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1324aeb8b3c247c2aa0844dd6c46cbbe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5adb5365efee4bd8ab33cb81420b7e12 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bda288af32147658a7a6d5587f3cb86 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6e38e80ba609d18e810f7cae3011762ebe9e38d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 08:25:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1fef3793f5b4f089969a746ba8be33972ef58f10","revisions":{"6125bcb702a94fcb496e2f1c1e291b83f795756f":{"kind":"REWORK","_number":1,"created":"2022-06-30 03:41:07.000000000","uploader":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"ref":"refs/changes/11/848211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-designate","ref":"refs/changes/11/848211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-designate refs/changes/11/848211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-designate refs/changes/11/848211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-designate refs/changes/11/848211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-designate refs/changes/11/848211/1"}}},"commit":{"parents":[{"commit":"fc616b48c072f1dd425ff53f39f2f9a4163c2468","subject":"Merge \"Add *.charm to gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate/commit/fc616b48c072f1dd425ff53f39f2f9a4163c2468"}]}],"author":{"name":"Francesco de Simone","email":"francesco.desimone@canonical.com","date":"2022-06-29 20:53:16.000000000","tz":0},"committer":{"name":"Francesco de Simone","email":"francesco.desimone@canonical.com","date":"2022-06-30 03:40:30.000000000","tz":0},"subject":"Disable designate-mdns service until the configuration has been rendered","message":"Disable designate-mdns service until the configuration has been rendered\n\nAs per the bug report, designate-mdns service appear to restart multiple time.\nThis patch disable the service until the configuration has been rendered.\n\nCloses-Bug: #1925233\nChange-Id: Ib9843bda9c6967ce50cec8d721079bd70f7f3040\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate/commit/6125bcb702a94fcb496e2f1c1e291b83f795756f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate/commit/6125bcb702a94fcb496e2f1c1e291b83f795756f"}]},"branch":"refs/heads/master"},"1fef3793f5b4f089969a746ba8be33972ef58f10":{"kind":"REWORK","_number":2,"created":"2022-07-18 13:36:35.000000000","uploader":{"_account_id":34959,"name":"Francesco De Simone","email":"francesco.desimone@canonical.com","username":"fdesi"},"ref":"refs/changes/11/848211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-designate","ref":"refs/changes/11/848211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-designate refs/changes/11/848211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-designate refs/changes/11/848211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-designate refs/changes/11/848211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-designate refs/changes/11/848211/2"}}},"commit":{"parents":[{"commit":"fc616b48c072f1dd425ff53f39f2f9a4163c2468","subject":"Merge \"Add *.charm to gitignore\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate/commit/fc616b48c072f1dd425ff53f39f2f9a4163c2468"}]}],"author":{"name":"Francesco de Simone","email":"francesco.desimone@canonical.com","date":"2022-06-29 20:53:16.000000000","tz":0},"committer":{"name":"fdesi","email":"francesco.desimone@canonical.com","date":"2022-07-18 13:35:00.000000000","tz":0},"subject":"Disable designate-mdns service until the configuration has been rendered","message":"Disable designate-mdns service until the configuration has been rendered\n\nAs per the bug report, designate-mdns service appear to restart multiple time.\nThis patch disable the service until the configuration has been rendered.\n\nCloses-Bug: #1925233\nChange-Id: Ib9843bda9c6967ce50cec8d721079bd70f7f3040\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate/commit/1fef3793f5b4f089969a746ba8be33972ef58f10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-designate/commit/1fef3793f5b4f089969a746ba8be33972ef58f10"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"}}]}],"submit_requirements":[]}
