)]}'
{"id":"openstack%2Fcharm-guide~836109","triplet_id":"openstack%2Fcharm-guide~master~I70a1d5208b97923c081b359af3208f4de65eb6ca","project":"openstack/charm-guide","branch":"master","topic":"add-spellchecker","attention_set":{},"removed_from_attention_set":{"8992":{"account":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"last_update":"2022-04-07 18:46:02.000000000","reason":"Billy Olsen replied on the change"},"30561":{"account":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"last_update":"2022-04-12 21:53:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I70a1d5208b97923c081b359af3208f4de65eb6ca","subject":"Add a spellchecker","status":"MERGED","created":"2022-04-01 01:02:45.000000000","updated":"2022-04-12 21:55:32.000000000","submitted":"2022-04-12 21:53:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836109-add-spellchecker","meta_rev_id":"e71bb5a43703bf8dbf40b806d9cdc2b055564c0d","_number":836109,"virtual_id_number":836109,"owner":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 21:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":2,"date":"2022-04-12 21:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"all":[{"value":1,"date":"2022-04-12 21:43:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 01:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 18:15:00.000000000","updated_by":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"reviewer":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"state":"REVIEWER"}],"messages":[{"id":"59fde797a17704a807fb441bd4684fff10dcb117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-01 01:02:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b38555c174a283a37d1859787f0f2641283a66e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 01:07:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4050864628704662b1e3e78903fc5800 : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"aa33780d56d855a438a701060fada6bc4969da81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-01 17:50:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ff285fa090cf142608d3ce6757e1462a98d4522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 17:58:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9d428b23ef485897f3c34a24b0072e : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a5e0c3ecc819c4dcab8fe94b574ba882fefaaab7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-01 18:12:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"36d451ab1fc3e232b80729a3364676bba9c132c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 18:17:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8215b4441904896bb5776ae51d73903 : FAILURE in 3m 25s","accounts_in_message":[],"_revision_number":3},{"id":"526abe79b0e30916fd1fbf8ed231a1a7e1a17568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-01 18:24:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4654d84ea0d9be6ac593728dd0729080ff75ebb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 18:31:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c65c1e7e69a433d9397dedc837ff258 : FAILURE in 5m 18s","accounts_in_message":[],"_revision_number":4},{"id":"2a442abcc0c2548cb850624a6683e9e712b8f2ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-01 18:35:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8a23dc97faef874fe4e31b96da8a4fb36d019be0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 18:39:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102572fb9929477a8712b76b3c457a9a : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":5},{"id":"2b1f2e6aff2217ddc3b0f2f17dfbce5dc50bbaaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-01 22:17:35.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"f13026f16687acf5fd557dc43da32aebfa6df9e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 22:21:47.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e743a64b7e9f4f09b35af6995ccc486e : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":6},{"id":"0f11df67798a59c5f9858b304ed27620dfe9c106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-07 18:21:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1585d0377703710fb7c3430e60380f1eee07af92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 18:31:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a2ff80ebcf459b9750c6678ae1d0ee : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":7},{"id":"90f8b2fd524f94fd54094eddc6bfbcb24e5086f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-07 18:35:00.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9c045529ee1325ed6c9953637188a7bf23c1e161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 18:43:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e93c6c194744860b50ed9c323e02eef : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":8},{"id":"637e46b3f77be2ba9b979cae31845e084a439557","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2022-04-07 18:46:02.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bddc73c670401ec57139518b92e04311de3dbf81","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2022-04-07 18:46:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"07163082cbd5ed6525370d05230f71cf4aa15427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-07 19:46:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"af57c1f68d75fb32f0ae301f3a8172a5e695b799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 19:51:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936818b07bd945b0aa409622c0c2162c : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":9},{"id":"10cfb4646dbc98c81211ccc3552bf7b5261a960e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-07 22:11:02.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"e7b53cc52f2d461c17fe4ad345b20c0339ae0596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 22:22:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f5e390ea3504e69aad6fa88fd17a7a2 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":10},{"id":"f07f287d16e15b019a92e0f7233b2371fa95c434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2022-04-08 05:54:31.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"09e5c69773aebd8a2ba9f6bcb306eb1380699576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 05:59:53.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433a29fc8b0b4a71a7273ae7315ecaf9 : SUCCESS in 3m 57s\n- charm-guide-spellcheck https://zuul.opendev.org/t/openstack/build/863fd922fc1442128601e988830162e3 : FAILURE in 3m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"925282d99c1d9f5cbd91a6583aa681ae4721e9b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2022-04-08 06:09:23.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"45efbc6189666f5075452610b1d79ddd3d666888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 06:14:47.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4bc93d3a99d45ecb6e6c14b8372bfeb : SUCCESS in 3m 45s\n- charm-guide-spellcheck https://zuul.opendev.org/t/openstack/build/a54a2a414f894e0ba939f742a0f83880 : FAILURE in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"66b49a684162bc5725fc27f23e062a8651921699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"date":"2022-04-08 06:22:26.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ff3ca4c7070577d551f2cca25afa59d066ae0b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 06:38:19.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50e13539040d423aacf3eb8951fa554a : SUCCESS in 3m 48s\n- charm-guide-spellcheck https://zuul.opendev.org/t/openstack/build/fec68d83d18341fb8aa5acaedacc3b2e : FAILURE in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"be37c1f999ad6e7d3b073aac45510da1a954e550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-08 15:22:16.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"2668332032424e6f2db9b50ae4697345bfacca3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 15:31:27.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98fab28bfc014de3ae57e064318ed902 : SUCCESS in 4m 08s\n- charm-guide-spellcheck https://zuul.opendev.org/t/openstack/build/597c5cc9e2004921a0e5f1976b671be9 : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c2cfc6b75b3d58f409d0d1fdea2522e0e12a044e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-08 16:09:19.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"432dea1e03b6b7405684d53ebe19adc4550daba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 16:18:29.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f82ed49e924a2ea092f694e594578e : SUCCESS in 3m 53s\n- charm-guide-spellcheck https://zuul.opendev.org/t/openstack/build/de325ef14a4b46e5a09ae0aa54bef3d1 : FAILURE in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5272e2c467b8443398dc546165d54b7099377b20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-11 16:06:22.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"24f06e674b2bd74b0a272f189ce7f8fd2fbdff07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 16:14:11.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8dfede5b1224394980e88aa38f579e5 : SUCCESS in 4m 23s\n- charm-guide-spellcheck https://zuul.opendev.org/t/openstack/build/81f3df04427a417483d58b369b75ebe0 : FAILURE in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"754a83851d3daf0c61dfe52d9ff02c7c1a72c029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"date":"2022-04-11 16:45:15.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"aa742afd3f65896859e52f87edd22de41e0ffff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 16:55:29.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb59f0c85d4c43a598c9ce11fce0c01e : SUCCESS in 5m 10s\n- charm-guide-spellcheck https://zuul.opendev.org/t/openstack/build/a7cfef7a042246b7ad0f575aaf308e2c : SUCCESS in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"568b90c466178f50ef97b32bb99206223a51f8ba","author":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"},"date":"2022-04-12 21:43:19.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"307e48558d4aa9bf6adde08cd65bfe2fba64e15f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 21:43:55.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"e253a69082cce2e87b9a51cf7ed4673acba59ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 21:53:04.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205c433f93b94bbb936af838dade60ec : SUCCESS in 4m 01s\n- charm-guide-spellcheck https://zuul.opendev.org/t/openstack/build/710d82d71f05433cb9de9a5e39f89951 : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9e1680776029aedd8984f919ab3d15bc0cc7663f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 21:53:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"e71bb5a43703bf8dbf40b806d9cdc2b055564c0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 21:55:32.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003eecab90c645fdb906f76d10dfaced : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"eb1e9e207b22a925cc84e4d4b492c787e1907798","revisions":{"e96101014759e0bc9ef3a273b80da13105bff192":{"kind":"REWORK","_number":1,"created":"2022-04-01 01:02:45.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/1"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:22.000000000","tz":-240},"subject":"Add spellchecker","message":"Add spellchecker\n\nTo conform to Canonical documentation standards add\na spellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd seeding files that extend the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors, and if\nwarnings are made to be non-critical then this will\naffect other aspects of a build, which is not desirable.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e96101014759e0bc9ef3a273b80da13105bff192"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e96101014759e0bc9ef3a273b80da13105bff192"}]},"branch":"refs/heads/master"},"1ea896ad89381c85d91affc81c7a51c589427724":{"kind":"REWORK","_number":2,"created":"2022-04-01 17:50:58.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/2"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 17:49:25.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to Canonical documentation standards add\na spellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors when the\nword used is correct, and if warnings are made to be\nnon-critical then this will affect other aspects of a\nbuild, which is not desirable.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/1ea896ad89381c85d91affc81c7a51c589427724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/1ea896ad89381c85d91affc81c7a51c589427724"}]},"branch":"refs/heads/master"},"5cd5fc8256a446dbd415d0d3be8b34793f160f50":{"kind":"REWORK","_number":3,"created":"2022-04-01 18:12:07.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/3"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 18:12:01.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to Canonical documentation standards add\na spellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors when the\nword used is correct, and if warnings are made to be\nnon-critical then this will affect other aspects of a\nbuild, which is not desirable.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/5cd5fc8256a446dbd415d0d3be8b34793f160f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/5cd5fc8256a446dbd415d0d3be8b34793f160f50"}]},"branch":"refs/heads/master"},"9c9a0314880b75d460c995c169c2c0f53c159a41":{"kind":"REWORK","_number":4,"created":"2022-04-01 18:24:00.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/4"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 18:23:33.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to Canonical documentation standards add\na spellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors when the\nword used is correct, and if warnings are made to be\nnon-critical then this will affect other aspects of a\nbuild, which is not desirable.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/9c9a0314880b75d460c995c169c2c0f53c159a41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/9c9a0314880b75d460c995c169c2c0f53c159a41"}]},"branch":"refs/heads/master"},"3ba130778d013a6fb0a31ff3e63b6388882fb308":{"kind":"REWORK","_number":5,"created":"2022-04-01 18:35:30.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/5"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 18:35:25.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to Canonical documentation standards add\na spellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors when the\nword used is correct, and if warnings are made to be\nnon-critical then this will affect other aspects of a\nbuild, which is not desirable.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/3ba130778d013a6fb0a31ff3e63b6388882fb308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/3ba130778d013a6fb0a31ff3e63b6388882fb308"}]},"branch":"refs/heads/master"},"46dd03cca8306f3999a8db25e4267d514fc84288":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-04-01 22:17:35.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/6"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 22:16:27.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors when the\nword used is correct, and if warnings are made to be\nnon-critical then this will affect other aspects of a\nbuild, which is not desirable.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/46dd03cca8306f3999a8db25e4267d514fc84288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/46dd03cca8306f3999a8db25e4267d514fc84288"}]},"branch":"refs/heads/master"},"cee74328dda4a94f41d9fe172124f41264f735f6":{"kind":"REWORK","_number":7,"created":"2022-04-07 18:21:24.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/7"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-07 18:21:12.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors when the\nword used is correct, and if warnings are made to be\nnon-critical then this will affect other aspects of a\nbuild, which is not desirable.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/cee74328dda4a94f41d9fe172124f41264f735f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/cee74328dda4a94f41d9fe172124f41264f735f6"}]},"branch":"refs/heads/master"},"e040aefae245c3c9ed903e5fdedd1dc51a925e92":{"kind":"REWORK","_number":8,"created":"2022-04-07 18:35:00.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/8"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-07 18:34:39.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors when the\nword used is correct, and if warnings are made to be\nnon-critical then this will affect other aspects of a\nbuild, which is not desirable.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e040aefae245c3c9ed903e5fdedd1dc51a925e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e040aefae245c3c9ed903e5fdedd1dc51a925e92"}]},"branch":"refs/heads/master"},"3d6314046aec5c6ab23bb1028d07b67fdf0a6f8d":{"kind":"REWORK","_number":9,"created":"2022-04-07 19:46:23.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/9"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-07 19:46:08.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a doc build. It is\nnon-obvious how to resolve spelling errors when the\nword used is correct, and if warnings are made to be\nnon-critical then this will affect other aspects of a\nbuild, which is not desirable.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/3d6314046aec5c6ab23bb1028d07b67fdf0a6f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/3d6314046aec5c6ab23bb1028d07b67fdf0a6f8d"}]},"branch":"refs/heads/master"},"ea70e1b78b806619c91a0781bb2b05a0ed66be4f":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-04-07 22:11:02.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/10"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-07 22:07:04.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a normal doc build; add\na new tox target.\n\nAdd a non-voting Zuul job that consumes the new tox\ntarget.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/ea70e1b78b806619c91a0781bb2b05a0ed66be4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/ea70e1b78b806619c91a0781bb2b05a0ed66be4f"}]},"branch":"refs/heads/master"},"645f7f82a8dd200b7754c88874d461c62ccd4a89":{"kind":"REWORK","_number":11,"created":"2022-04-08 05:54:31.000000000","uploader":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"ref":"refs/changes/09/836109/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/11"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Chris MacNaughton","email":"chris@centauri.solutions","date":"2022-04-08 05:54:30.000000000","tz":120},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a normal doc build; add\na new tox target.\n\nAdd a non-voting Zuul job that consumes the new tox\ntarget.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/645f7f82a8dd200b7754c88874d461c62ccd4a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/645f7f82a8dd200b7754c88874d461c62ccd4a89"}]},"branch":"refs/heads/master"},"65f6f30030baa7931936270675bad4122615e1ad":{"kind":"REWORK","_number":12,"created":"2022-04-08 06:09:23.000000000","uploader":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"ref":"refs/changes/09/836109/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/12"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Chris MacNaughton","email":"chris@centauri.solutions","date":"2022-04-08 06:09:18.000000000","tz":120},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a normal doc build; add\na new tox target.\n\nAdd a non-voting Zuul job that consumes the new tox\ntarget.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/65f6f30030baa7931936270675bad4122615e1ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/65f6f30030baa7931936270675bad4122615e1ad"}]},"branch":"refs/heads/master"},"c719e20ca8a1c014e7f8cd6f77c64be5e26c1504":{"kind":"REWORK","_number":13,"created":"2022-04-08 06:22:26.000000000","uploader":{"_account_id":20634,"name":"Chris MacNaughton","email":"chris.macnaughton@canonical.com","username":"Chris.MacNaughton"},"ref":"refs/changes/09/836109/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/13"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Chris MacNaughton","email":"chris@centauri.solutions","date":"2022-04-08 06:22:21.000000000","tz":120},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a normal doc build; add\na new tox target.\n\nAdd a non-voting Zuul job that consumes the new tox\ntarget.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/c719e20ca8a1c014e7f8cd6f77c64be5e26c1504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/c719e20ca8a1c014e7f8cd6f77c64be5e26c1504"}]},"branch":"refs/heads/master"},"ccdca3c248373e5caea07f5fbefb5a19ceff3d6f":{"kind":"REWORK","_number":14,"created":"2022-04-08 15:22:16.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/14"}}},"commit":{"parents":[{"commit":"e8de78b7c84693d0443fd02a06e688ec6a6503b8","subject":"Add release notes for `scheduler-max-attempts`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/e8de78b7c84693d0443fd02a06e688ec6a6503b8"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-08 15:22:09.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a normal doc build; add\na new tox target.\n\nAdd a non-voting Zuul job that consumes the new tox\ntarget.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/ccdca3c248373e5caea07f5fbefb5a19ceff3d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/ccdca3c248373e5caea07f5fbefb5a19ceff3d6f"}]},"branch":"refs/heads/master"},"a2eb455634741b660375f30cbce4f412a6339ba0":{"kind":"REWORK","_number":15,"created":"2022-04-08 16:09:19.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/15"}}},"commit":{"parents":[{"commit":"7086480bf85f7411b5f3d96efbafe22dc380f7cf","subject":"Merge \"Add vTPM to Charm Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/7086480bf85f7411b5f3d96efbafe22dc380f7cf"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-08 15:57:53.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a normal doc build; add\na new tox target.\n\nAdd a non-voting Zuul job that consumes the new tox\ntarget.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nFuture work is necessary in terms of making checking more\nintelligent. As such, file \u0027dubious_words.txt\u0027 has been\nadded to temporarily store those words that should be\nfiltered. For instance, the word \u0027tis\u0027 is in this file\nbecause it is part of proper noun \u0027tpm-tis\u0027. Hyphenated\nwords (or words in single quotes) could be exempt from\nthe check.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/a2eb455634741b660375f30cbce4f412a6339ba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/a2eb455634741b660375f30cbce4f412a6339ba0"}]},"branch":"refs/heads/master"},"de9597245e2702412f4ef4b4e784acf23e4bccda":{"kind":"REWORK","_number":16,"created":"2022-04-11 16:06:22.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/16"}}},"commit":{"parents":[{"commit":"7086480bf85f7411b5f3d96efbafe22dc380f7cf","subject":"Merge \"Add vTPM to Charm Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/7086480bf85f7411b5f3d96efbafe22dc380f7cf"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-11 16:05:59.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a normal doc build; add\na new tox target.\n\nAdd a non-voting Zuul job that consumes the new tox\ntarget.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nFuture work is necessary in terms of making checking more\nintelligent. As such, file \u0027dubious_words.txt\u0027 has been\nadded to temporarily store those words that should be\nfiltered. For instance, the word \u0027tis\u0027 is in this file\nbecause it is part of proper noun \u0027tpm-tis\u0027. Hyphenated\nwords (or words in single quotes) could be exempt from\nthe check.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/de9597245e2702412f4ef4b4e784acf23e4bccda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/de9597245e2702412f4ef4b4e784acf23e4bccda"}]},"branch":"refs/heads/master"},"eb1e9e207b22a925cc84e4d4b492c787e1907798":{"kind":"REWORK","_number":17,"created":"2022-04-11 16:45:15.000000000","uploader":{"_account_id":30561,"name":"Peter Matulis","email":"peter.matulis@canonical.com","username":"pmatulis"},"ref":"refs/changes/09/836109/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-guide","ref":"refs/changes/09/836109/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-guide refs/changes/09/836109/17"}}},"commit":{"parents":[{"commit":"7086480bf85f7411b5f3d96efbafe22dc380f7cf","subject":"Merge \"Add vTPM to Charm Guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/7086480bf85f7411b5f3d96efbafe22dc380f7cf"}]}],"author":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-01 00:54:19.000000000","tz":-240},"committer":{"name":"Peter Matulis","email":"peter.matulis@canonical.com","date":"2022-04-11 16:45:10.000000000","tz":-240},"subject":"Add a spellchecker","message":"Add a spellchecker\n\nTo conform to documentation best practices add a\nspellchecker. A Sphinx extension is used.\n\nCorrect any current spelling mistakes across the\ndoc set.\n\nAdd a seeding file that extends the system dictionary.\n\nGoing forward, the \u0027spelling_words.txt\u0027 file will\nbe used to extend the dictionary.\n\nDo not enforce spelling during a normal doc build; add\na new tox target.\n\nAdd a non-voting Zuul job that consumes the new tox\ntarget.\n\nThe doc-contrib documentation will include information\non the spellchecker in a subsequent PR.\n\nFuture work is necessary in terms of making checking more\nintelligent. As such, file \u0027dubious_words.txt\u0027 has been\nadded to temporarily store those words that should be\nfiltered. For instance, the word \u0027tis\u0027 is in this file\nbecause it is part of proper noun \u0027tpm-tis\u0027. Hyphenated\nwords (or words in single quotes) could be exempt from\nthe check.\n\nChange-Id: I70a1d5208b97923c081b359af3208f4de65eb6ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/eb1e9e207b22a925cc84e4d4b492c787e1907798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-guide/commit/eb1e9e207b22a925cc84e4d4b492c787e1907798"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8992,"name":"Billy Olsen","email":"billy.olsen@canonical.com","username":"billy-olsen"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
