)]}'
{"id":"openstack%2Fcharm-keystone~712040","triplet_id":"openstack%2Fcharm-keystone~master~I23940720c24527ee34149f035c3bdf9ff54812c9","project":"openstack/charm-keystone","branch":"master","topic":"use-bootstrap","hashtags":[],"change_id":"I23940720c24527ee34149f035c3bdf9ff54812c9","subject":"Replace use of admin_token with Keystone bootstrap","status":"MERGED","created":"2020-03-10 10:50:51.000000000","updated":"2020-03-18 16:43:44.000000000","submitted":"2020-03-18 16:43:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712040-1584549824942-9c168db8","meta_rev_id":"77803141a52f62a7eb40dea699dff8393d2c5649","_number":712040,"virtual_id_number":712040,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-18 16:06:02.000000000","_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":1,"date":"2020-03-14 15:03:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-18 16:43:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"all":[{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":2,"date":"2020-03-17 13:36:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":1,"date":"2020-03-18 16:29:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-14 15:03:53.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-17 13:36:00.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2020-03-18 16:29:43.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"},{"updated":"2020-03-18 16:43:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc69ba59aba7d44fca82b027b8a873236e9ed075","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-10 10:50:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20a56cf7636b4968585f5ec101619b663ea5a2e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 11:04:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0acfee27f37c4d0d91c9e93dca28e80e : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0104d5ce265f4386b6db3764a89c0961 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c19cd2decef74fd481f5df6de34dc95d : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/226b77907748465ba7b1c08c332631cc : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50315106594048e7a0fd0d714b38e1c4 : FAILURE in 9m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a48d369721e430d87a8d850121aa642 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"551db4a6a45960502890c622097ba6ce6884b125","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2020-03-10 11:35:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/1/15104/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/1/15104/index.html : FAILURE\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/1/15104/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/1/15104/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":1},{"id":"e9b719ea010a553a26e1c00fb146427502d9afad","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-10 15:04:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02a7e6ac7b4ab92564acca8e72fdb9c266e6506e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 15:18:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c0f86e7e66d4ce2b0a7a0b2645bea25 : SUCCESS in 4m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88e1da79b2f44ef7afb99d5d0b8a00c8 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e4d8da14b1f495aa259f2e81198ff5b : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f192acbfcbbc48ab9b0f819fd2cc6c14 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b635d4292fd74cfd862d414ae1ceaa3f : FAILURE in 8m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4514a8d5892a497083f2d341f77be241 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":2},{"id":"21faadde544007ba3e10ba366dbb9743064db04d","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2020-03-10 15:52:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/2/15109/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/2/15109/index.html : FAILURE\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/2/15109/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/2/15109/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":2},{"id":"21c5b1d1a50d438218bcff1e7560e8f010698931","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-11 13:12:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6529b8da715f47ab02381de83d1c7ca8acdf99a","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-11 13:35:40.000000000","message":"Topic changed from bug/1859844 to use-bootstrap","accounts_in_message":[],"_revision_number":3},{"id":"c69e20ddd5524f1f5d66a463019617d04c30e748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 14:11:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f16889f4a044579b1314c857dde7cc7 : SUCCESS in 4m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7fb25f4e8d44f3980875b35443082a4 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9405ca0693f433c8f1391deec1d2d6d : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d305275ced444d6b9ce287731f968aa : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c729651afd5440ffad30397602c55746 : FAILURE in 10m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/537ecdc5023040ec93d8a16268ab38cb : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"abab68a700b3c9116f290424ccdcef9a1c71f9b9","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2020-03-11 14:24:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/3/15175/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/3/15175/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/3/15175/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-keystone/712040/3/15175/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":3},{"id":"92ed810f61b5eb939d7809e61065081fa697376f","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-11 14:26:32.000000000","message":"Patch Set 3:\n\ncharm-recheck-full","accounts_in_message":[],"_revision_number":3},{"id":"1cc838fd7282e5698f80f3a0de1a900915fd62f1","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2020-03-11 21:28:07.000000000","message":"Patch Set 3:\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_func_full https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-keystone/712040/3/4969/index.html : SUCCESS\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-keystone/712040/3/4969/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-keystone/712040/3/4969/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-keystone/712040/3/4969/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":3},{"id":"7718be98c4373c3844b02eacda25748bd24658e6","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-13 08:52:50.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"b08b1ceeec9b42efdf33bc1167e7a53b6544e953","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-13 09:22:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"43716237711242432213a96715e2c06e0e38e33b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 09:35:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ba5895b21c40ab8f72831182ba3351 : SUCCESS in 3m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9e9931de390b4b38b0929ee8500bb23f : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7aba9ecf3f7442cb87de145b375b019 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53f509df5f304d118b7992d6884da6d9 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07162a7853a24946a625f0c67fb2afcc : FAILURE in 7m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b90775a014ed4d60aa2737efab65a896 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":4},{"id":"9a15ce4dd7862157914c35882125c3129379218d","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-14 08:33:53.000000000","message":"Patch Set 4:\n\ncharm-recheck-full","accounts_in_message":[],"_revision_number":4},{"id":"0bca8391d57a14920a78292f45996115f0acec18","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2020-03-14 15:03:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_func_full https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-keystone/712040/4/5031/index.html : SUCCESS\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-keystone/712040/4/5031/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-keystone/712040/4/5031/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-keystone/712040/4/5031/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":4},{"id":"bdb08fc0667974d1b0f76ba5024c72b1d1066f5c","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-16 10:06:13.000000000","message":"Topic changed from bug/1859844 to use-bootstrap","accounts_in_message":[],"_revision_number":4},{"id":"1b648426eff5d55498feda7fb9cd60a45c36abcc","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2020-03-17 13:36:00.000000000","message":"Patch Set 4: Code-Review+2\n\nI\u0027m generally +2 on this change but its a large review so would appreciate a second set of eye\u0027s over it to confirm its good toland.","accounts_in_message":[],"_revision_number":4},{"id":"cd54e26fc31106b3124d80956686e064373a51f6","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2020-03-18 15:56:50.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nJust a couple of minor comments; we could merge as is and pick them up in another commit?","accounts_in_message":[],"_revision_number":4},{"id":"a55e010321d745b0f272682d36b6762d60abdb2d","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2020-03-18 16:06:02.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b9dcf7821b932e428cceac1d3c14f2a421134e0c","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2020-03-18 16:29:43.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4e1573eaaeb13c0213492a592d01cfda646f35a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 16:30:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"db7fb04ae1abac1883bf4eca71b082d936bc8322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 16:43:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e86c69aca8d4b2e8032daa9af29b336 : SUCCESS in 5m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36bac709005a43018f7a3d8ffde264b7 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/533996928286427790ea8736da684a0a : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22fa59c4b74b4fdcaa4950fc45e3da43 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":4},{"id":"fc663c32b4d41a7ad38c1d00aa8625f1473d87d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 16:43:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0a02c30fe5f4650235519897b71588ae22fa0971","revisions":{"bb5c169a52a3cb0a64c67a21c487d2c779fbc0cc":{"kind":"REWORK","_number":1,"created":"2020-03-10 10:50:51.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/40/712040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/40/712040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/1"}}},"commit":{"parents":[{"commit":"3ea039b3eb30155086d2de0a95607a9b7a22d97c","subject":"Unpin flake8 and fix lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/3ea039b3eb30155086d2de0a95607a9b7a22d97c"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2020-03-09 14:06:09.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2020-03-10 10:47:40.000000000","tz":60},"subject":"[WIP] Replace use of admin_token with Keystone bootstrap","message":"[WIP] Replace use of admin_token with Keystone bootstrap\n\nStop the use of the admin_token and use the bootstrap process\nto initialize Keystone instead.  Fortunately the implementation\nof the bootstrap process is both idempotent when it needs to be\nand it can be safely called on an existing deployment.\n\nSubsequently we can migrate by just removing the admin_token\nfrom the configuration and create new credentials for use by\nthe charm with a call to ``keystone-manage bootstrap``.\n\nRemove configuration templates for versions prior to Mitaka, by\ndoing this we need to move any configuration initially defined\nprior to Miataka forward to the ``templates/mitaka`` folder.\n\nA side effect of this migration is that newly bootstrapped\ndeployments will get their ``default`` domain created with a\nliteral ID of ``default``.  Prior to this change third party\nsoftware making assumptions about that being the case may have\nhad issues.\n\nChange-Id: I23940720c24527ee34149f035c3bdf9ff54812c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/bb5c169a52a3cb0a64c67a21c487d2c779fbc0cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/bb5c169a52a3cb0a64c67a21c487d2c779fbc0cc"}]},"branch":"refs/heads/master"},"44f649ebeb82217adb8033a8b1b4a7030416cbdb":{"kind":"REWORK","_number":2,"created":"2020-03-10 15:04:36.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/40/712040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/40/712040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/2"}}},"commit":{"parents":[{"commit":"3765c0b123bcd2f87ee04fe17e89e236928fc6d6","subject":"Unpin flake8 and fix lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/3765c0b123bcd2f87ee04fe17e89e236928fc6d6"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2020-03-09 14:06:09.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2020-03-10 15:03:09.000000000","tz":60},"subject":"[WIP] Replace use of admin_token with Keystone bootstrap","message":"[WIP] Replace use of admin_token with Keystone bootstrap\n\nStop the use of the admin_token and use the bootstrap process\nto initialize Keystone instead.  Fortunately the implementation\nof the bootstrap process is both idempotent when it needs to be\nand it can be safely called on an existing deployment.\n\nSubsequently we can migrate by just removing the admin_token\nfrom the configuration and create new credentials for use by\nthe charm with a call to ``keystone-manage bootstrap``.\n\nRemove configuration templates for versions prior to Mitaka, by\ndoing this we need to move any configuration initially defined\nprior to Miataka forward to the ``templates/mitaka`` folder.\n\nA side effect of this migration is that newly bootstrapped\ndeployments will get their ``default`` domain created with a\nliteral ID of ``default``.  Prior to this change third party\nsoftware making assumptions about that being the case may have\nhad issues.\n\nChange-Id: I23940720c24527ee34149f035c3bdf9ff54812c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/44f649ebeb82217adb8033a8b1b4a7030416cbdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/44f649ebeb82217adb8033a8b1b4a7030416cbdb"}]},"branch":"refs/heads/master"},"a480287338d6a490750c4a9423a2a622df902bd2":{"kind":"REWORK","_number":3,"created":"2020-03-11 13:12:40.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/40/712040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/40/712040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/3"}}},"commit":{"parents":[{"commit":"3765c0b123bcd2f87ee04fe17e89e236928fc6d6","subject":"Unpin flake8 and fix lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/3765c0b123bcd2f87ee04fe17e89e236928fc6d6"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2020-03-09 14:06:09.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2020-03-11 12:54:33.000000000","tz":60},"subject":"Replace use of admin_token with Keystone bootstrap","message":"Replace use of admin_token with Keystone bootstrap\n\nStop the use of the admin_token and use the bootstrap process\nto initialize Keystone instead.  Fortunately the implementation\nof the bootstrap process is both idempotent when it needs to be\nand it can be safely called on an existing deployment.\n\nSubsequently we can migrate by just removing the admin_token\nfrom the configuration and create new credentials for use by\nthe charm with a call to ``keystone-manage bootstrap``.\n\nRemove configuration templates for versions prior to Mitaka, by\ndoing this we need to move any configuration initially defined\nprior to Miataka forward to the ``templates/mitaka`` folder.\n\nA side effect of this migration is that newly bootstrapped\ndeployments will get their ``default`` domain created with a\nliteral ID of ``default``.  Prior to this change third party\nsoftware making assumptions about that being the case may have\nhad issues.\n\nCloses-Bug: #1859844\nCloses-Bug: #1837113\nRelated-Bug: #1774733\nCloses-Bug: #1648719\nCloses-Bug: #1578678\nFunc-Test-Pr: https://github.com/openstack-charmers/zaza-openstack-tests/pull/191\nChange-Id: I23940720c24527ee34149f035c3bdf9ff54812c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/a480287338d6a490750c4a9423a2a622df902bd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/a480287338d6a490750c4a9423a2a622df902bd2"}]},"branch":"refs/heads/master"},"0a02c30fe5f4650235519897b71588ae22fa0971":{"kind":"REWORK","_number":4,"created":"2020-03-13 09:22:39.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/40/712040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/40/712040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/40/712040/4"}}},"commit":{"parents":[{"commit":"3765c0b123bcd2f87ee04fe17e89e236928fc6d6","subject":"Unpin flake8 and fix lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/3765c0b123bcd2f87ee04fe17e89e236928fc6d6"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2020-03-09 14:06:09.000000000","tz":60},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2020-03-13 08:52:10.000000000","tz":60},"subject":"Replace use of admin_token with Keystone bootstrap","message":"Replace use of admin_token with Keystone bootstrap\n\nStop the use of the admin_token and use the bootstrap process\nto initialize Keystone instead.  Fortunately the implementation\nof the bootstrap process is both idempotent when it needs to be\nand it can be safely called on an existing deployment.\n\nSubsequently we can migrate by just removing the admin_token\nfrom the configuration and create new credentials for use by\nthe charm with a call to ``keystone-manage bootstrap``.\n\nRemove configuration templates for versions prior to Mitaka, by\ndoing this we need to move any configuration initially defined\nprior to Miataka forward to the ``templates/mitaka`` folder.\n\nA side effect of this migration is that newly bootstrapped\ndeployments will get their ``default`` domain created with a\nliteral ID of ``default``.  Prior to this change third party\nsoftware making assumptions about that being the case may have\nhad issues.\n\nCloses-Bug: #1859844\nCloses-Bug: #1837113\nRelated-Bug: #1774733\nCloses-Bug: #1648719\nCloses-Bug: #1578678\nFunc-Test-Pr: https://github.com/openstack-charmers/zaza-openstack-tests/pull/191\nChange-Id: I23940720c24527ee34149f035c3bdf9ff54812c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/0a02c30fe5f4650235519897b71588ae22fa0971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/0a02c30fe5f4650235519897b71588ae22fa0971"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
