)]}'
{"id":"openstack%2Fcharm-keystone~990774","triplet_id":"openstack%2Fcharm-keystone~stable%2F2024.1~If9bb2d926001f7266b66eb37fac9ee4d07920a43","project":"openstack/charm-keystone","branch":"stable/2024.1","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2026-06-02 11:24:12.000000000","reason":"\u003cGERRIT_ACCOUNT_6737\u003e replied on the change","reason_account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},"5112":{"account":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"last_update":"2026-06-02 11:37:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If9bb2d926001f7266b66eb37fac9ee4d07920a43","subject":"Cleanup Caracal CI","status":"MERGED","created":"2026-05-30 00:46:39.000000000","updated":"2026-06-02 11:37:59.000000000","submitted":"2026-06-02 11:37:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990774-openstack-charms-ci-cleanup","meta_rev_id":"f6b5148bcf33140be11935031723408538423ab2","_number":990774,"virtual_id_number":990774,"owner":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 11:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-02 11:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":2,"date":"2026-06-01 23:24:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-02 11:24:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-30 01:01:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 23:24:10.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-06-02 11:24:12.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"ad277d2fec1fde59719f578bced18904c338ef47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-05-30 00:46:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e7d6df0199e6c57feacb4451be30473a248e5fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 01:01:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/475c891a3b794648a5c117b90b6c3ae8\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/9de715ee167345e8baef3e2ccd6c063f : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac4b03d0888c4b4ca48cdea9dacb866e : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/411921a7f16548a9aaa9e282d4c568e1 : SUCCESS in 4m 31s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/3535ffe7ea3647ca85169bc725909dd1 : SUCCESS in 6m 10s\n- charmbuild https://zuul.opendev.org/t/openstack/build/6f1141ee7d7a42ffbba1dfd191a50d58 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8eb6a629e31f31f2b8bb1dc704d9f15d63b9f489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-01 17:47:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9abdcac9d8bfcd34a8a0f20f037fb9c3371f5dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 18:02:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42f9728f7c8c4f9f8f2560dd9027adb6\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/be89abda3fd44d84996f60373aa10839 : SUCCESS in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a4bbbe88c034e7dada6cf3b85671020 : SUCCESS in 8m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/911c100439884b3796f61ba3b285b557 : SUCCESS in 3m 00s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/f969eca5060a4b06aa00da0de1399128 : SUCCESS in 4m 02s\n- charmbuild https://zuul.opendev.org/t/openstack/build/c5348ba952504f039bab57c9fa276330 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":2},{"id":"3b24c2cfbce9feee7f3e451a13f8dbe931d8b624","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-06-01 23:24:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e53e8f1d6f84bf9095fb1e63d686a81f1411237c","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-02 11:24:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d94a1b65fe071bde899275f77ff3d2d6e43c50e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:24:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"851e19e121d76220ae7a6c2dc7aedd2d21977840","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:37:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/753298016f3b491ebb24f5200731ff23\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/0aa7448ae47e44619a4640fcd3ed01cc : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93e366c2f2804175b32b018a71c0f835 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c1d9451eb714466b1c11caf4036728e : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f6b5148bcf33140be11935031723408538423ab2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 11:37:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"55203a6a29b47bd253a0d3ddbf2af1e03d3bc7e4","revisions":{"0e26771c5e286a837c07b689da563e3f58a36752":{"kind":"REWORK","_number":1,"created":"2026-05-30 00:46:39.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/74/990774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/74/990774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/74/990774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/74/990774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/74/990774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/74/990774/1"}}},"commit":{"parents":[{"commit":"7581d77900c1b63f06cdf1c9d0bad13bc99f681f","subject":"Charm Helpers Sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/7581d77900c1b63f06cdf1c9d0bad13bc99f681f"}]}],"author":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-05-29 09:10:37.000000000","tz":0},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-05-29 09:10:37.000000000","tz":0},"subject":"Cleanup Caracal CI","message":"Cleanup Caracal CI\n\n  * generate merged-requirements-py310.txt\n  * use juju36 test constraints from zaza\n  * add update-requirement env to tox.ini\n  * tox.ini:\n    - update to python3 (py310)\n  * test-requirements.in:\n    - generate test-requirements-py310.txt\n  * update test bundles to work with charmcraft v3\n\nChange-Id: If9bb2d926001f7266b66eb37fac9ee4d07920a43\nSigned-off-by: Seyeong Kim \u003cseyeong.kim@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/0e26771c5e286a837c07b689da563e3f58a36752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/0e26771c5e286a837c07b689da563e3f58a36752"}]},"branch":"refs/heads/stable/2024.1"},"55203a6a29b47bd253a0d3ddbf2af1e03d3bc7e4":{"kind":"REWORK","_number":2,"created":"2026-06-01 17:47:43.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/74/990774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/74/990774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/74/990774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/74/990774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/74/990774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/74/990774/2"}}},"commit":{"parents":[{"commit":"7581d77900c1b63f06cdf1c9d0bad13bc99f681f","subject":"Charm Helpers Sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/7581d77900c1b63f06cdf1c9d0bad13bc99f681f"}]}],"author":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-05-29 09:10:37.000000000","tz":0},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-01 17:47:22.000000000","tz":60},"subject":"Cleanup Caracal CI","message":"Cleanup Caracal CI\n\n  * generate merged-requirements-py310.txt\n  * use juju36 test constraints from zaza\n  * add update-requirement env to tox.ini\n  * tox.ini:\n    - update to python3 (py310)\n  * test-requirements.in:\n    - generate test-requirements-py310.txt\n  * update test bundles to work with charmcraft v3\n\nChange-Id: If9bb2d926001f7266b66eb37fac9ee4d07920a43\nSigned-off-by: Seyeong Kim \u003cseyeong.kim@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/55203a6a29b47bd253a0d3ddbf2af1e03d3bc7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/55203a6a29b47bd253a0d3ddbf2af1e03d3bc7e4"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
