)]}'
{"id":"openstack%2Fcharm-keystone~990777","triplet_id":"openstack%2Fcharm-keystone~stable%2Fzed~I0b3494be1f653c84471ffb42aed2fe6e2945e8c6","project":"openstack/charm-keystone","branch":"stable/zed","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2026-06-02 19:34:51.000000000","reason":"\u003cGERRIT_ACCOUNT_6737\u003e replied on the change","reason_account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},"5112":{"account":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"last_update":"2026-06-02 19:44:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b3494be1f653c84471ffb42aed2fe6e2945e8c6","subject":"Cleanup Zed CI","status":"MERGED","created":"2026-05-30 01:21:19.000000000","updated":"2026-06-02 19:44:38.000000000","submitted":"2026-06-02 19:44:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990777-openstack-charms-ci-cleanup","meta_rev_id":"1acb1957b790f4c439fc6eb2221f3dbcaec2b84d","_number":990777,"virtual_id_number":990777,"owner":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 19:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-02 19:34:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":2,"date":"2026-06-01 19:51:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-02 19:34:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-30 01:27:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 19:51:42.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-06-02 19:34:51.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"a1627953d9b2720a44d2dbbfdb3fb0d5425a0302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-05-30 01:21:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dfd92fcc2d5d4a40dcecb14f8a195c0d5f112ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 01:27:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5e059cf40fd4c7bb917899ce9402cd5\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f48be1a315894240844cabee31dd49c5 : SUCCESS in 2m 12s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/e7fbab4fcffd4bd294500578d0cb7192 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c3acded45de4692139afcacade915b5f566c7989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-01 18:09:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"756cfeb957ab4a1de3e781988c7863b2326d5def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 18:30:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f49051333bbf494594edbc34f840d621\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c6a55c231cc46e1b9c5187a5dd1771d : SUCCESS in 2m 39s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/b49ac77cc09a42a3a58bc9aa687f358c : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":2},{"id":"7e98e00506266338c66b26e1be36b68f160ba3f4","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-06-01 19:51:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8103bbbc3ec65d823a8923c6ec1ba61dc961a7c","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-02 19:34:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"824ea43520160df462431c55be522b7971061b04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:35:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8cc17db8a13421bb8d3e51e6a7cc90ce4e41f051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:44:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2c799a38904de582caadf0101b862e\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a625245f03f460e9bbfcd5ea000eaa0 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1acb1957b790f4c439fc6eb2221f3dbcaec2b84d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:44:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1fa0a38b49b1444eae165aa5ac477a91706372bd","revisions":{"66f9a22a9b550fff5cff19e3fc6d05b9a6869c43":{"kind":"REWORK","_number":1,"created":"2026-05-30 01:21:19.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/77/990777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/77/990777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/77/990777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/77/990777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/77/990777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/77/990777/1"}}},"commit":{"parents":[{"commit":"7f89bd899813a399531cae1942345ac952dde814","subject":"Sync charmhelpers and pin MarkupSafe\u003c3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/7f89bd899813a399531cae1942345ac952dde814"}]}],"author":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-05-29 09:22:21.000000000","tz":0},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-05-30 01:19:03.000000000","tz":0},"subject":"Cleanup Zed CI","message":"Cleanup Zed CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py310 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n* .zuul.yaml\n   - switch to openstack-python3-charm-jobs-zed\n* fix:\n   - pep8 E721/E275 in keystone_context.py and keystone_utils.py\n\nChange-Id: I0b3494be1f653c84471ffb42aed2fe6e2945e8c6\nSigned-off-by: Seyeong Kim \u003cseyeong.kim@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/66f9a22a9b550fff5cff19e3fc6d05b9a6869c43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/66f9a22a9b550fff5cff19e3fc6d05b9a6869c43"}]},"branch":"refs/heads/stable/zed"},"1fa0a38b49b1444eae165aa5ac477a91706372bd":{"kind":"REWORK","_number":2,"created":"2026-06-01 18:09:55.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/77/990777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/77/990777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/77/990777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/77/990777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/77/990777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/77/990777/2"}}},"commit":{"parents":[{"commit":"7f89bd899813a399531cae1942345ac952dde814","subject":"Sync charmhelpers and pin MarkupSafe\u003c3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/7f89bd899813a399531cae1942345ac952dde814"}]}],"author":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-05-29 09:22:21.000000000","tz":0},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-01 18:05:08.000000000","tz":60},"subject":"Cleanup Zed CI","message":"Cleanup Zed CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py310 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n* .zuul.yaml\n   - switch to openstack-python3-charm-jobs-zed\n* fix:\n   - pep8 E721/E275 in keystone_context.py and keystone_utils.py\n\nChange-Id: I0b3494be1f653c84471ffb42aed2fe6e2945e8c6\nSigned-off-by: Seyeong Kim \u003cseyeong.kim@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/1fa0a38b49b1444eae165aa5ac477a91706372bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/1fa0a38b49b1444eae165aa5ac477a91706372bd"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
