)]}'
{"id":"openstack%2Fcharm-keystone~991637","triplet_id":"openstack%2Fcharm-keystone~stable%2Fyoga~I81ccc1377161392a282686e45e194e881e2e5b8a","project":"openstack/charm-keystone","branch":"stable/yoga","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"38780":{"account":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"last_update":"2026-06-04 17:07:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81ccc1377161392a282686e45e194e881e2e5b8a","subject":"Update tox.ini and re-generate requirements","status":"MERGED","created":"2026-06-04 13:32:49.000000000","updated":"2026-06-04 17:07:38.000000000","submitted":"2026-06-04 17:07:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991637-openstack-charms-ci-cleanup","meta_rev_id":"82136abbb946cd3e4b23919e2feb90ffd00fa082","_number":991637,"virtual_id_number":991637,"owner":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-04 17:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2026-06-04 16:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2026-06-04 16:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 13:39:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-04 16:57:52.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"}],"messages":[{"id":"56b996a4d915ffdc2f080573959749128dc39e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-04 13:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03568e54dd2fe36033d3894195e2b23d6945cb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 13:39:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4066a9cfb9af48ff81b419be7ce131cd\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/1eab45c17bac494ca1b2cfe45f03caae : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb2324682a2e4420967423902efc4e40 : SUCCESS in 5m 14s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/4aea8713c46d4fc28c5344b6ec2705c9 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9cb1a884a14ea51ccca581bc6adbe0a6e93d6bcf","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-06-04 16:57:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82a2e7c5536af8e3210edbe78c84ed0d78e36826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 16:58:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd53d5e05a006252fe138602f9edd3ac3a5e5c4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:07:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d0f6fe0a9c4579888d619794f86571\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/a26655b5ae3b4d57aecfa31b2f399e7c : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce1765a7e275456d85ed63877a380b2c : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"82136abbb946cd3e4b23919e2feb90ffd00fa082","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:07:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc30864ebf25b9fcbe557136032984e5f79fc1f1","revisions":{"fc30864ebf25b9fcbe557136032984e5f79fc1f1":{"kind":"REWORK","_number":1,"created":"2026-06-04 13:32:49.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/37/991637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-keystone","ref":"refs/changes/37/991637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/37/991637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/37/991637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-keystone refs/changes/37/991637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-keystone refs/changes/37/991637/1"}}},"commit":{"parents":[{"commit":"6c02e009df0bd990fbcde3cb66667f7c33577380","subject":"Cleanup Yoga CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/6c02e009df0bd990fbcde3cb66667f7c33577380"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 13:32:06.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 13:32:06.000000000","tz":300},"subject":"Update tox.ini and re-generate requirements","message":"Update tox.ini and re-generate requirements\n\ntox.ini needs to have env for specific Python versions as this\nis needed by the CI Zuul jobs, e.g., py310, py312 etc. Since we\nare using the same CI cleanup procedure as with other charms,\nthis may also modify requirements files, so we re-generate\nthem to be sure they are in sync.\n\nChange-Id: I81ccc1377161392a282686e45e194e881e2e5b8a\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/fc30864ebf25b9fcbe557136032984e5f79fc1f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-keystone/commit/fc30864ebf25b9fcbe557136032984e5f79fc1f1"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
