)]}'
{"id":"openstack%2Fcharm-neutron-api~897423","triplet_id":"openstack%2Fcharm-neutron-api~master~Ib8e861377c9f691f31de0f707823dbf8ac0fde52","project":"openstack/charm-neutron-api","branch":"master","topic":"bug/2036953","attention_set":{},"removed_from_attention_set":{"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2023-10-05 20:03:27.000000000","reason":"Change was submitted"},"11805":{"account":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"last_update":"2023-10-05 20:03:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8e861377c9f691f31de0f707823dbf8ac0fde52","subject":"Drop hyperv mechanism driver and package","status":"MERGED","created":"2023-10-05 12:10:25.000000000","updated":"2023-10-06 00:23:50.000000000","submitted":"2023-10-05 20:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"897423-bug/2036953","meta_rev_id":"ed153bc42b839638f9bcad9c7e73483a57d874a2","_number":897423,"virtual_id_number":897423,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-10-05 17:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-05 20:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-05 19:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":1,"date":"2023-10-06 00:23:50.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-05 19:53:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-05 12:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 13:09:36.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-05 13:55:43.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2023-10-05 14:17:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-05 19:53:46.000000000","updated_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"reviewer":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"state":"REVIEWER"}],"messages":[{"id":"e0e5ee5703c1c98b4b5fb7974ec801f22a99e24f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-05 12:10:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10744f48a162a893ac3e55d759202ba810042048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-05 12:11:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f3b391c78fb476b6f367f118e603fe59051672b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 12:21:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71b0761b954240699d73d100c85f69f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6316ebd0b1a14acd84288fee34d2e92d : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9240a0a5644442a294aee83790e2496f : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c76e338197524d9caacfd6c78f8ea066 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb95d13ad4934f8a95f771004f4a1448 : FAILURE in 4m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2d1de3d8418415c986107b264602bad : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"517b0a097d455083521a9c3b45eb52fd1adee436","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-10-05 13:09:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/163/897423/2/check/charm-build/1633efc/ : SUCCESS in 9m 37s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/f07/897423/2/check/osci-lint/f07d418/ : SUCCESS in 6m 11s\n- mantic-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/b83/897423/2/check/mantic-bobcat/b832d5c/ : SUCCESS in 41m 48s (non-voting)\n- jammy-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/fef/897423/2/check/jammy-bobcat/fef9be3/ : SUCCESS in 47m 32s\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/877/897423/2/check/lunar-antelope/877f2b0/ : SUCCESS in 47m 38s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/758/897423/2/check/jammy-antelope/75854be/ : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":2},{"id":"222c7e821016a01a30fcca20458471db35a8c5b7","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2023-10-05 13:55:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5f30c4fe44375beed5c7a895c4afe7177c4db53b","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-05 15:46:21.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ebc930908e0d10f60a6aa0c59eac930c142f8c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-05 15:51:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f3fde92cbd624d0ac3dddd4e4d4dbb5fae035d18","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-05 15:57:56.000000000","message":"Topic set to bug/2036953","accounts_in_message":[],"_revision_number":3},{"id":"08da5493e3dad7fad4d5c1dd66f19e8dfca33717","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 16:04:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6335856109194ce2ac3929d137e52790\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da12a4ea28dc4ff09095468413270139 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc7cd8d39f2b4864b6fb30b7ce13498e : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6414bb8e1b3147da9543a887dd8f860d : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b142f5e8be546539244b8872e012c58 : FAILURE in 4m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8e15839a73a48e2a7e0934c45a58245 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":3},{"id":"0da1a668cabb58633cd67da4e03cb92073bfc0bb","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2023-10-05 17:55:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- charm-build https://openstack-ci-reports.ubuntu.com/artifacts/54b/897423/3/check/charm-build/54b5e02/ : SUCCESS in 8m 21s\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/a54/897423/3/check/osci-lint/a54b219/ : SUCCESS in 6m 26s\n- mantic-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/c2e/897423/3/check/mantic-bobcat/c2e73c7/ : SUCCESS in 1h 02m 38s (non-voting)\n- jammy-bobcat https://openstack-ci-reports.ubuntu.com/artifacts/1ad/897423/3/check/jammy-bobcat/1adecd8/ : SUCCESS in 53m 08s\n- lunar-antelope https://openstack-ci-reports.ubuntu.com/artifacts/65f/897423/3/check/lunar-antelope/65fe9f1/ : SUCCESS in 55m 47s (non-voting)\n- jammy-antelope https://openstack-ci-reports.ubuntu.com/artifacts/69e/897423/3/check/jammy-antelope/69e6760/ : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":3},{"id":"23969e015edc7a6bce5aedb5334c9cb605b631ce","author":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"},"date":"2023-10-05 19:53:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b263720d4fbcbb4dba7d319c6b038ec17a9c1fed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 19:54:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0f2d3e04868c3e4cd89b59eb5e6284f7441e8576","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 20:03:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5b32ace13641e0aebb9ac08ec3784a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/240ef88281804d92ab968f1328f70bf1 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edf7bc0515ca41be98f4a6cdfc8e8443 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4567b08566bc45a7b24328e66fedc27f : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"6b60f9f9078e7fac73411fac36d678e787c47fdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 20:03:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ed153bc42b839638f9bcad9c7e73483a57d874a2","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2023-10-06 00:23:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a6c5567354175755762ec9a3e20fc628a06fcd43","revisions":{"32447ca7fd02329bccbc6dbe42deabed8d904c05":{"kind":"REWORK","_number":1,"created":"2023-10-05 12:10:25.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/23/897423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/23/897423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/1"}}},"commit":{"parents":[{"commit":"7a0925af7b58df1bee987df0e8c9faa171934986","subject":"Ensure get_requests_for_local_unit doesn\u0027t fail on incomplete relation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/7a0925af7b58df1bee987df0e8c9faa171934986"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-05 11:52:21.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-05 12:10:24.000000000","tz":-240},"subject":"Drop hyperv mechanism driver and package","message":"Drop hyperv mechanism driver and package\n\nThe networking-hyperv project is no longer maintained and\nhas been removed from Ubuntu in Mantic.\n\nChange-Id: Ib8e861377c9f691f31de0f707823dbf8ac0fde52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/32447ca7fd02329bccbc6dbe42deabed8d904c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/32447ca7fd02329bccbc6dbe42deabed8d904c05"}]},"branch":"refs/heads/master"},"59b2185a58a1aa1d6114522a05e6a47c5470088e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-05 12:11:05.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/23/897423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/23/897423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/2"}}},"commit":{"parents":[{"commit":"7a0925af7b58df1bee987df0e8c9faa171934986","subject":"Ensure get_requests_for_local_unit doesn\u0027t fail on incomplete relation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/7a0925af7b58df1bee987df0e8c9faa171934986"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-05 11:52:21.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-05 12:10:49.000000000","tz":-240},"subject":"Drop hyperv mechanism driver and package","message":"Drop hyperv mechanism driver and package\n\nThe networking-hyperv project is no longer maintained and\nhas been removed from Ubuntu in Mantic.\n\nCloses-Bug: #2036953\n\nChange-Id: Ib8e861377c9f691f31de0f707823dbf8ac0fde52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/59b2185a58a1aa1d6114522a05e6a47c5470088e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/59b2185a58a1aa1d6114522a05e6a47c5470088e"}]},"branch":"refs/heads/master"},"a6c5567354175755762ec9a3e20fc628a06fcd43":{"kind":"REWORK","_number":3,"created":"2023-10-05 15:51:22.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/23/897423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/23/897423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/23/897423/3"}}},"commit":{"parents":[{"commit":"7a0925af7b58df1bee987df0e8c9faa171934986","subject":"Ensure get_requests_for_local_unit doesn\u0027t fail on incomplete relation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/7a0925af7b58df1bee987df0e8c9faa171934986"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-05 11:52:21.000000000","tz":0},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-05 15:51:17.000000000","tz":-240},"subject":"Drop hyperv mechanism driver and package","message":"Drop hyperv mechanism driver and package\n\nThe networking-hyperv project is no longer maintained and\nhas been removed from Ubuntu in Mantic.\n\nCloses-Bug: #2036953\n\nChange-Id: Ib8e861377c9f691f31de0f707823dbf8ac0fde52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/a6c5567354175755762ec9a3e20fc628a06fcd43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/a6c5567354175755762ec9a3e20fc628a06fcd43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":2424,"name":"Felipe Reyes","email":"felipe.reyes@canonical.com","username":"freyes"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
