)]}'
{"id":"openstack%2Fcharm-neutron-api~990883","triplet_id":"openstack%2Fcharm-neutron-api~stable%2Fzed~I6e4de817b1075dec24983b0025debc4b473166a2","project":"openstack/charm-neutron-api","branch":"stable/zed","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2026-06-02 19:40:30.000000000","reason":"\u003cGERRIT_ACCOUNT_6737\u003e replied on the change","reason_account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},"38780":{"account":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"last_update":"2026-06-02 19:45:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e4de817b1075dec24983b0025debc4b473166a2","subject":"Cleanup Zed CI","status":"MERGED","created":"2026-06-01 07:08:24.000000000","updated":"2026-06-02 19:45:57.000000000","submitted":"2026-06-02 19:45:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990883-openstack-charms-ci-cleanup","meta_rev_id":"d11cbe1d7fcdfdca5606956c4d9b06c55adfd5ad","_number":990883,"virtual_id_number":990883,"owner":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 19:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-02 19:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":2,"date":"2026-06-01 23:34:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-02 19:40:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-01 07:27:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 23:34:15.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-06-02 19:40:30.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"3d8f566a969f6d8a8c6fad0fe34e5ccaf1df2742","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-01 07:08:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e182b892d0b0a5f1b42d30bd3cbb992f3a1cc7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 07:27:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1afae2f7b6e944f5a3ad37409f891c59\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/022077a1f24248a9861dd6ecb8b73a18 : SUCCESS in 5m 28s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/4b887ccfc57c46e6b58af401e491ed1b : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1474649f499fc2d647b4f51c55d987dd38c2bad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-01 07:31:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7217134c3ec5c9593aa2c837b54b28b2bcd88522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 07:36:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2acd35f33b54682885c67f5ce50e11b\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21fd83aae7ae4871ae7a210b03bb6ab6 : SUCCESS in 3m 20s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/90b6a91decf94881ad211d10db833931 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a960974da09b75c1536cf4f03ed033146b2bb948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-01 16:23:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c7875ec8265d9110611dea1135427a87d9d9714f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:28:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1805c4e66e2b4919b1597adc8e33c8d0\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2def086d263041ebbb3d361ed416893c : SUCCESS in 2m 41s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/0d43f92683214b2695bd9e5009433cda : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"bf1cb067b7ce7cf7393d6337e92bfbb917290839","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-06-01 23:34:15.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b1cd94221f5014d4dd7269b3afaa3d1676b6da2d","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-02 19:40:30.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5a723daf5eae314f5f1a56ad8057b77b001d85d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:41:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8b765eb48eb5a16e8431aafa164ee2a145d49cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:45:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c263002cfd047d2b3e31cbff48a6eda\n\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ef8014e2b6640c88fc7843ff8f8807c : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d11cbe1d7fcdfdca5606956c4d9b06c55adfd5ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 19:45:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9a64881b6523ee268a2e0d707341c41149aa3fee","revisions":{"52c025ae1b9236a87249e5de16da8b367575badc":{"kind":"REWORK","_number":1,"created":"2026-06-01 07:08:24.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/83/990883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/83/990883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/1"}}},"commit":{"parents":[{"commit":"461a03de75fa7c092a8cf1812a0f4ac10c79f10c","subject":"Add database pool size/overflow config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/461a03de75fa7c092a8cf1812a0f4ac10c79f10c"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:07:22.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:08:09.000000000","tz":300},"subject":"Cleanup Zed CI","message":"Cleanup Zed CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n* .zuul.yaml:\n   - switch to charm jobs to get py310 support\n\nChange-Id: I6e4de817b1075dec24983b0025debc4b473166a2\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/52c025ae1b9236a87249e5de16da8b367575badc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/52c025ae1b9236a87249e5de16da8b367575badc"}]},"branch":"refs/heads/stable/zed"},"881c2ea89cfca8efd41dfc5c999ae683bb59424f":{"kind":"REWORK","_number":2,"created":"2026-06-01 07:31:12.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/83/990883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/83/990883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/2"}}},"commit":{"parents":[{"commit":"461a03de75fa7c092a8cf1812a0f4ac10c79f10c","subject":"Add database pool size/overflow config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/461a03de75fa7c092a8cf1812a0f4ac10c79f10c"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:07:22.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:30:33.000000000","tz":300},"subject":"Cleanup Zed CI","message":"Cleanup Zed CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n  - update-requirements env to tox.ini\n* .zuul.yaml:\n  - switch to charm jobs to get py310 support\n* hooks/neutron_api_utils.py:\n  - fix whitespace issue in migrate_neutron_database()\n\nChange-Id: I6e4de817b1075dec24983b0025debc4b473166a2\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/881c2ea89cfca8efd41dfc5c999ae683bb59424f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/881c2ea89cfca8efd41dfc5c999ae683bb59424f"}]},"branch":"refs/heads/stable/zed"},"9a64881b6523ee268a2e0d707341c41149aa3fee":{"kind":"REWORK","_number":3,"created":"2026-06-01 16:23:47.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/83/990883/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/83/990883/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/83/990883/3"}}},"commit":{"parents":[{"commit":"461a03de75fa7c092a8cf1812a0f4ac10c79f10c","subject":"Add database pool size/overflow config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/461a03de75fa7c092a8cf1812a0f4ac10c79f10c"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:07:22.000000000","tz":300},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-01 16:23:38.000000000","tz":60},"subject":"Cleanup Zed CI","message":"Cleanup Zed CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - test-requirements-py310.txt\n* add:\n  - update-requirements env to tox.ini\n* .zuul.yaml:\n  - switch to charm jobs to get py310 support\n* hooks/neutron_api_utils.py:\n  - fix whitespace issue in migrate_neutron_database()\n\nChange-Id: I6e4de817b1075dec24983b0025debc4b473166a2\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/9a64881b6523ee268a2e0d707341c41149aa3fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/9a64881b6523ee268a2e0d707341c41149aa3fee"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
