)]}'
{"id":"openstack%2Fcharm-neutron-api~990884","triplet_id":"openstack%2Fcharm-neutron-api~stable%2Fyoga~I71b29bbc9969e1b2cda322b3acb8edd4cd5e8ede","project":"openstack/charm-neutron-api","branch":"stable/yoga","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"6737":{"account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"last_update":"2026-06-02 20:53:08.000000000","reason":"\u003cGERRIT_ACCOUNT_6737\u003e replied on the change","reason_account":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},"38780":{"account":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"last_update":"2026-06-02 21:02:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71b29bbc9969e1b2cda322b3acb8edd4cd5e8ede","subject":"Cleanup Yoga CI","status":"MERGED","created":"2026-06-01 07:15:43.000000000","updated":"2026-06-02 21:02:56.000000000","submitted":"2026-06-02 21:02:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990884-openstack-charms-ci-cleanup","meta_rev_id":"5c05f452f0901bca13fcb3506cd6bfa61c7c98fd","_number":990884,"virtual_id_number":990884,"owner":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 21:02:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-02 20:53:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":2,"date":"2026-06-01 23:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-02 20:53:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-01 07:24:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 23:25:03.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2026-06-02 20:53:08.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"4ada2441706f2771ace6b88125864bc664a4453f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-01 07:15:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa4114081a85863ece1294eddca3fe2cec9ce6ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 07:24:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b8560efc67a4c9592d3aff01af46335\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/e4a2e16dd11b4493bfff39536d42c951 : FAILURE in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/661e36fad9504efcb4654fa89d04ac1f : SUCCESS in 3m 49s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/dd9dae8c97c14ac0aa2cdbe2498555ac : SUCCESS in 2m 20s\n\nWarning:\n  Comments left for invalid file hooks/neutron_api_utils.py","accounts_in_message":[],"_revision_number":1},{"id":"fd53342a7725d8e252d33ec0060450c6cf9cb576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-01 07:31:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"0934c4677367c24ffe9267662b59922f83fb2fc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 07:41:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3df4b0a4275d42b692b79327c032270f\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/b5aee9ceec0043c7aa86718a1bbc894b : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/865ee72361ea4953940f70da01f24111 : SUCCESS in 2m 41s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/c6c1a3737bdf4f9d961aee864f90cada : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d9ee1123adccfcba6eaae85a5d66cccdd01cb70b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-01 16:24:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d3760fe625aaeb2651ade2ad822a603053e97d9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-01 16:37:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97ae47537e244841b13657927f801c84\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/81b42e0c7d59480eb6082d231f3520fa : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd0aaba37b294b6db5c6c332c52c4348 : SUCCESS in 3m 50s\n- openstack-tox-cover-charm-jammy https://zuul.opendev.org/t/openstack/build/cbd0431f0afa424bb10de481208849bc : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"1960033b64283473daab73f86958fb7a4e41b9c3","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2026-06-01 23:25:03.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d9a4e9913cd732db6cba8f369ace374ddf351b87","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-02 20:53:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b5fbe81a481f56644551a4ebce42299e9fbe1d11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 20:54:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6e601328198f9bfd0c94ca3f70e256ad9ef40d48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 21:02:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a557fc9b5d9473583c908f0d598715f\n\n- openstack-tox-pep8-charm-jammy https://zuul.opendev.org/t/openstack/build/075937806977473da2ef3ebc6983818d : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02c06c1444e0459c9d3beda42dec70be : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":3},{"id":"5c05f452f0901bca13fcb3506cd6bfa61c7c98fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 21:02:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b27797a5846f07c41bf9cc8c42a6892aa6a45b46","revisions":{"43fae9959b864b4ed81100061f8d01a891547a91":{"kind":"REWORK","_number":1,"created":"2026-06-01 07:15:43.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/84/990884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/84/990884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/1"}}},"commit":{"parents":[{"commit":"6a3465e0cefa28994916f63339214ef1a176bb3f","subject":"Add database pool size/overflow config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/6a3465e0cefa28994916f63339214ef1a176bb3f"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:15:22.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:15:22.000000000","tz":300},"subject":"Cleanup Yoga CI","message":"Cleanup Yoga CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - merged-requirements-py38.txt\n  - test-requirements-py310.txt\n* add:\n   - update-requirements env to tox.ini\n   - update-requirements-focal env to tox.ini\n\nChange-Id: I71b29bbc9969e1b2cda322b3acb8edd4cd5e8ede\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/43fae9959b864b4ed81100061f8d01a891547a91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/43fae9959b864b4ed81100061f8d01a891547a91"}]},"branch":"refs/heads/stable/yoga"},"caac48caa90bb58f727951e18dc58deb7efaf5ee":{"kind":"REWORK","_number":2,"created":"2026-06-01 07:31:51.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/84/990884/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/84/990884/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/2"}}},"commit":{"parents":[{"commit":"6a3465e0cefa28994916f63339214ef1a176bb3f","subject":"Add database pool size/overflow config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/6a3465e0cefa28994916f63339214ef1a176bb3f"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:15:22.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:31:33.000000000","tz":300},"subject":"Cleanup Yoga CI","message":"Cleanup Yoga CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - merged-requirements-py38.txt\n  - test-requirements-py310.txt\n* add:\n  - update-requirements env to tox.ini\n  - update-requirements-focal env to tox.ini\n* hooks/neutron_api_utils.py:\n  - fix whitespace issue in migrate_neutron_database()\n\nChange-Id: I71b29bbc9969e1b2cda322b3acb8edd4cd5e8ede\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/caac48caa90bb58f727951e18dc58deb7efaf5ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/caac48caa90bb58f727951e18dc58deb7efaf5ee"}]},"branch":"refs/heads/stable/yoga"},"b27797a5846f07c41bf9cc8c42a6892aa6a45b46":{"kind":"REWORK","_number":3,"created":"2026-06-01 16:24:13.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/84/990884/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/84/990884/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/84/990884/3"}}},"commit":{"parents":[{"commit":"6a3465e0cefa28994916f63339214ef1a176bb3f","subject":"Add database pool size/overflow config options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/6a3465e0cefa28994916f63339214ef1a176bb3f"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-01 07:15:22.000000000","tz":300},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-01 16:24:04.000000000","tz":60},"subject":"Cleanup Yoga CI","message":"Cleanup Yoga CI\n\n* update:\n  - charmcraft.yaml for v3\n  - tox.ini to use py3.10 by default\n  - test-requirements.in to use zaza for juju3\n  - update test bundles to work with charmcraft v3\n* generate:\n  - merged-requirements-py310.txt\n  - merged-requirements-py38.txt\n  - test-requirements-py310.txt\n* add:\n  - update-requirements env to tox.ini\n  - update-requirements-focal env to tox.ini\n* hooks/neutron_api_utils.py:\n  - fix whitespace issue in migrate_neutron_database()\n\nChange-Id: I71b29bbc9969e1b2cda322b3acb8edd4cd5e8ede\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/b27797a5846f07c41bf9cc8c42a6892aa6a45b46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/b27797a5846f07c41bf9cc8c42a6892aa6a45b46"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
