)]}'
{"id":"openstack%2Fcharm-neutron-api~991640","triplet_id":"openstack%2Fcharm-neutron-api~master~I17ab9bd1dacb8ca1a78def8452b7625b22f354f0","project":"openstack/charm-neutron-api","branch":"master","topic":"openstack-charms-ci-cleanup","attention_set":{},"removed_from_attention_set":{"38780":{"account":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"last_update":"2026-06-05 11:51:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I17ab9bd1dacb8ca1a78def8452b7625b22f354f0","subject":"Update tox.ini and re-generate requirements","status":"MERGED","created":"2026-06-04 13:36:41.000000000","updated":"2026-06-05 11:51:34.000000000","submitted":"2026-06-05 11:51:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991640-openstack-charms-ci-cleanup","meta_rev_id":"577b3335ef8bd6dba3b17b9e2926173303939d43","_number":991640,"virtual_id_number":991640,"owner":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-05 11:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":2,"date":"2026-06-05 11:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"all":[{"value":1,"date":"2026-06-05 11:41:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 13:46:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-05 11:41:13.000000000","updated_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"}],"messages":[{"id":"31d0b72595a59d32c5c8cba3a18c0c458febf9a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-04 13:36:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91eb3ebe68545274eb060e2cd6c8e2b087ea7c05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 13:46:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e8521b5f3ea4af2833db5a5c5716499\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b28164933e0425ca1ab9b45d2bd9842 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/015c0cadcb61485bbee600ed5b56f520 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/504bb5f4bea64216aa92861e4930b0b6 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"707e76598671bd1895f4051b096bdbab107578f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"date":"2026-06-04 17:19:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"82a8fcbf9394dd1eaf106ac3e32cc8ab303132f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:24:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c0b76a1c599462ba245bb6f89a355d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326fcd36b02d48008e43da45150fb70a : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f942d52524ac4b5a9cb95e57841bfba7 : SUCCESS in 3m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/150b6edffb1c4db8be2c0845fdc28117 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"3b3de132d8a4411693a55b2fb9605e606ee2b4fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-05 09:17:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"990adc220f39401891daa4caafdbdb6d8cbba247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 09:27:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf0d68c99c24f0590b4e13b97028e80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa83d3cd8ace43f89457603eaa6ca1c7 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/23258b6f11744082b6a5e7aeae8b2d40 : SUCCESS in 4m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf15353d31134c788e55e8dc0c5e677e : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"a010ceb98f9dce77b075087206f70da8e07c93c2","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2026-06-05 11:41:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db2e0aabdbf5f07d292268cc35aa6d96a4b42200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:41:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0075d69d53f8b38f6327caedfccd43a07376022d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:51:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46144c171dc94d0fb6b3e84e1ba318fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcabb8cc2a3e4fd2a39273c649484616 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da067484999a42ce8b71232ee6248328 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"577b3335ef8bd6dba3b17b9e2926173303939d43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 11:51:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4102c080ca61388a6b9208c118b9c00701a5cda6","revisions":{"0cc63d97751ac809eb07a0c472da5ea4d35734f4":{"kind":"REWORK","_number":1,"created":"2026-06-04 13:36:41.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/40/991640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/40/991640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/1"}}},"commit":{"parents":[{"commit":"4ed3442907895b600415db7bad1508c21f30f292","subject":"Add *_proxy envvars to tox passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/4ed3442907895b600415db7bad1508c21f30f292"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 13:36:23.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 13:36:23.000000000","tz":300},"subject":"Update tox.ini and re-generate requirements","message":"Update tox.ini and re-generate requirements\n\ntox.ini needs to have env for specific Python versions as this\nis needed by the CI Zuul jobs, e.g., py310, py312 etc. Since we\nare using the same CI cleanup procedure as with other charms,\nthis may also modify requirements files, so we re-generate\nthem to be sure they are in sync.\n\nChange-Id: I17ab9bd1dacb8ca1a78def8452b7625b22f354f0\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/0cc63d97751ac809eb07a0c472da5ea4d35734f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/0cc63d97751ac809eb07a0c472da5ea4d35734f4"}]},"branch":"refs/heads/master"},"9cbb2e9a2bbd6325fde239af9d87a90516740d8a":{"kind":"REWORK","_number":2,"created":"2026-06-04 17:19:51.000000000","uploader":{"_account_id":38780,"name":"Munir Siddiqui","display_name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","username":"munirsid"},"ref":"refs/changes/40/991640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/40/991640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/2"}}},"commit":{"parents":[{"commit":"4ed3442907895b600415db7bad1508c21f30f292","subject":"Add *_proxy envvars to tox passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/4ed3442907895b600415db7bad1508c21f30f292"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 17:19:22.000000000","tz":300},"committer":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 17:19:22.000000000","tz":300},"subject":"Update tox.ini and re-generate requirements","message":"Update tox.ini and re-generate requirements\n\ntox.ini needs to have env for specific Python versions as this\nis needed by the CI Zuul jobs, e.g., py310, py312 etc. Since we\nare using the same CI cleanup procedure as with other charms,\nthis may also modify requirements files, so we re-generate\nthem to be sure they are in sync.\n\nChange-Id: I17ab9bd1dacb8ca1a78def8452b7625b22f354f0\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/9cbb2e9a2bbd6325fde239af9d87a90516740d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/9cbb2e9a2bbd6325fde239af9d87a90516740d8a"}]},"branch":"refs/heads/master"},"4102c080ca61388a6b9208c118b9c00701a5cda6":{"kind":"REWORK","_number":3,"created":"2026-06-05 09:17:08.000000000","uploader":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"ref":"refs/changes/40/991640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-api","ref":"refs/changes/40/991640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-api refs/changes/40/991640/3"}}},"commit":{"parents":[{"commit":"4ed3442907895b600415db7bad1508c21f30f292","subject":"Add *_proxy envvars to tox passenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/4ed3442907895b600415db7bad1508c21f30f292"}]}],"author":{"name":"Munir Siddiqui","email":"munir.siddiqui@canonical.com","date":"2026-06-04 17:19:22.000000000","tz":300},"committer":{"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","date":"2026-06-05 09:16:31.000000000","tz":60},"subject":"Update tox.ini and re-generate requirements","message":"Update tox.ini and re-generate requirements\n\ntox.ini needs to have env for specific Python versions as this\nis needed by the CI Zuul jobs, e.g., py310, py312 etc. Since we\nare using the same CI cleanup procedure as with other charms,\nthis may also modify requirements files, so we re-generate\nthem to be sure they are in sync. Also adds HOME to tox passenv\nas it is needed for zaza/functests.\n\nChange-Id: I17ab9bd1dacb8ca1a78def8452b7625b22f354f0\nSigned-off-by: Munir Siddiqui \u003cmunir.siddiqui@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/4102c080ca61388a6b9208c118b9c00701a5cda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-api/commit/4102c080ca61388a6b9208c118b9c00701a5cda6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
