)]}'
{"id":"openstack%2Fcharm-neutron-openvswitch~778932","triplet_id":"openstack%2Fcharm-neutron-openvswitch~master~I07a7c57f55822d3fa0e24133cf6e166d12945ebe","project":"openstack/charm-neutron-openvswitch","branch":"master","topic":"wallaby-bundles","hashtags":[],"change_id":"I07a7c57f55822d3fa0e24133cf6e166d12945ebe","subject":"Test bundles for focal-wallaby and hirsute-wallaby","status":"ABANDONED","created":"2021-03-05 14:11:49.000000000","updated":"2021-10-05 18:03:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f4946ec816f3801812427e473a7585c1f146768","_number":778932,"virtual_id_number":778932,"owner":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2021-07-06 17:56:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-06 12:21:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-07-12 11:01:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"}],"CC":[{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-05 14:21:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 16:25:28.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 13:52:19.000000000","updated_by":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"reviewer":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"state":"CC"},{"updated":"2021-04-20 18:23:35.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"CC"},{"updated":"2021-05-06 10:18:42.000000000","updated_by":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"reviewer":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"state":"REVIEWER"}],"messages":[{"id":"6cbe75041491a84b956f6c92da778234df2ad76e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-03-05 14:11:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"073c9e8f7aa89d15e4d3ca959771e89a0e9ecffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 14:21:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd24a847241542f5b8eff389462121e5 : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/81299a90b9944294ae722e3e27bca72f : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59d3f6057d374e3da2b5318a447ed63d : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c13535bdbd904a09b13511010e945288 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/491c58357f304472b43bd55a3a748eb2 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01533a8dfa0f4b98a950aace53f55d1e : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"83ea85bfeb6aec1aad220e95756d1017a1d9fad3","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-03-05 16:25:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/1/22414/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/1/22414/index.html : FAILURE\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/1/22414/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":1},{"id":"fbf3524459a65ce436ee777b30c4c7928bf53dd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-03-15 12:45:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f57c880f2379189c697137895afed2b45bf2ef86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 12:53:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54021231f8e4c58b6c52d66a5ab5c06 : SUCCESS in 4m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0e8172ae1b6b437093048d145b47b198 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3801009cba074d54b3981c829993160e : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98cb2cbe0af84fbb945750056d56acdb : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09d4108e4955454bb0b7270427defa62 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e54bc2223c5e46238c1d092a94ce5393 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"dc2538731a86890fed00f361bcfaf75361f2d751","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-03-15 15:50:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/2/22597/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/2/22597/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/2/22597/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/2/22597/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":2},{"id":"1cbcce4e0559abc9be01ccaf778ada1fb946309a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-03-23 13:15:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d3c32ecc898c68fa4e7b80947b14756952dda7cb","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-03-23 13:25:28.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/415/778932/3/check/osci-lint/4154fe0/index.html : SUCCESS in 7m 00s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/1dd/778932/3/check/tox-py35/1dda9bf/index.html : FAILURE in 4m 08s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/75d/778932/3/check/tox-py36/75d48cc/index.html : SUCCESS in 6m 39s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/85d/778932/3/check/tox-py37/85dd215/index.html : SUCCESS in 5m 57s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/bda/778932/3/check/tox-py38/bda2ec2/index.html : SUCCESS in 6m 08s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat migrate-ovn-focal-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED\n- migrate-ovn-focal-wallaby-dvr-snat migrate-ovn-focal-wallaby-dvr-snat : SKIPPED\n- migrate-ovn-hirsute-wallaby-dvr-snat migrate-ovn-hirsute-wallaby-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":3},{"id":"5690a441aad38c439b0c48d9bf1f3c5f9aef5204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:27:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4099dc6f8d954e58be37c9ea464337b1 : SUCCESS in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cd644d48d5804dc5931590aa5a6470a8 : FAILURE in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c842fe982a7445b5a96a7bcaf7d0180b : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b37f066b679a4ffdb35702fa3c30c775 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe6385b3be014f198ea482b1d8beca02 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0bd0cf916694a8f98bec99511c6d1b5 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":3},{"id":"64563739808d02b45d9d8969bd74eada9798db54","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-03-23 14:27:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/3/22799/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/3/22799/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/3/22799/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/3/22799/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":3},{"id":"3033a340d8c1d08ef0bbf09e6b01117364931f8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-03-30 17:29:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"619aabb549d8eaf36e78621ed087cf85c99127af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 18:28:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/450078b37d764dbbb6db385453fc7f8c : SUCCESS in 6m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e830c565cd3b4619aa30eb67983350e5 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e125a685dc4f4d53b4c41a5e4a0752f2 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52aed2a87b3e44adadf0b325da442079 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f9da325cb864d46ae62705be0fb9b79 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4e26bef6e2847dda23dd7d589a1a17a : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":4},{"id":"a4ddeab38af01e648de917e4fc84600ea2a416f9","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-03-30 19:31:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/4/23007/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/4/23007/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/4/23007/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/4/23007/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":4},{"id":"4365ffdaaed3257b99603676a48532a8173547a0","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-03-31 16:20:39.000000000","message":"Patch Set 4:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/c4b/778932/4/check/osci-lint/c4b7d86/index.html : SUCCESS in 11m 50s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/91a/778932/4/check/tox-py35/91a12d8/index.html : SUCCESS in 11m 19s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/70c/778932/4/check/tox-py36/70cc941/index.html : SUCCESS in 10m 10s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/d59/778932/4/check/tox-py37/d5925a5/index.html : SUCCESS in 17m 53s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/5a1/778932/4/check/tox-py38/5a1a133/index.html : SUCCESS in 11m 49s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/6d8/778932/4/check/xenial-mitaka_ovs/6d8a5b7/index.html : SUCCESS in 55m 56s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d42/778932/4/check/bionic-queens-dvr-snat/d4248b4/index.html : SUCCESS in 54m 45s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/092/778932/4/check/bionic-stein-dvr-snat/092d8d4/index.html : SUCCESS in 1h 00m 06s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/01e/778932/4/check/bionic-train-dvr-snat/01e15fb/index.html : SUCCESS in 1h 04m 45s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/0b2/778932/4/check/bionic-ussuri-dvr-snat/0b2d79f/index.html : SUCCESS in 1h 03m 35s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/bca/778932/4/check/migrate-ovn-focal-ussuri-dvr-snat/bca0c44/index.html : SUCCESS in 1h 07m 28s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/451/778932/4/check/migrate-ovn-focal-victoria-dvr-snat/4511fec/index.html : SUCCESS in 1h 13m 39s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/eab/778932/4/check/migrate-ovn-focal-wallaby-dvr-snat/eab2629/index.html : FAILURE in 27m 29s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/b0f/778932/4/check/migrate-ovn-hirsute-wallaby-dvr-snat/b0fee97/index.html : FAILURE in 19m 46s","accounts_in_message":[],"_revision_number":4},{"id":"c383aff8067eb9a292bcca71b30f14b2693fc4c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-04-10 15:09:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f40ac932edfd205384675010c507700c03ad2bfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 15:20:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ccdb5e97e147658d8d5e1b770227e1 : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a0d19796b3745d396ed728649fe8a60 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a00e0303bcac4f1e895b5e5d7cd44051 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba1794778f9041e6bf6bbfca73c318c2 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6b90a79771e4f02bd1e73eba36a5fa4 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/398c05c547b0405eb35afe2ac60269ad : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":5},{"id":"a35bdd2bb7197cf4135378185947a45a06f56b75","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-10 16:58:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/5/23326/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/5/23326/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/5/23326/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/5/23326/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":5},{"id":"c0acbc6150745930126c652e54e363b315ee6166","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-10 18:53:07.000000000","message":"Patch Set 5:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/f05/778932/5/check/osci-lint/f05ce12/index.html : SUCCESS in 6m 27s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/c35/778932/5/check/tox-py35/c35f62a/index.html : SUCCESS in 6m 57s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/582/778932/5/check/tox-py36/5826efb/index.html : SUCCESS in 7m 58s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/323/778932/5/check/tox-py37/323dbc6/index.html : SUCCESS in 7m 39s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/7b7/778932/5/check/tox-py38/7b72f61/index.html : SUCCESS in 6m 50s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/931/778932/5/check/xenial-mitaka_ovs/9310cf7/index.html : SUCCESS in 50m 00s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/947/778932/5/check/bionic-queens-dvr-snat/9472b08/index.html : SUCCESS in 1h 22m 36s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/f0e/778932/5/check/bionic-stein-dvr-snat/f0e6f9e/index.html : SUCCESS in 1h 20m 48s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/61b/778932/5/check/bionic-train-dvr-snat/61b53d1/index.html : SUCCESS in 1h 21m 30s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d93/778932/5/check/bionic-ussuri-dvr-snat/d93052e/index.html : SUCCESS in 1h 29m 06s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/641/778932/5/check/migrate-ovn-focal-ussuri-dvr-snat/6416bb2/index.html : SUCCESS in 1h 27m 24s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ef7/778932/5/check/migrate-ovn-focal-victoria-dvr-snat/ef76bd4/index.html : SUCCESS in 1h 15m 09s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/1a8/778932/5/check/migrate-ovn-focal-wallaby-dvr-snat/1a84d92/index.html : FAILURE in 1h 39m 09s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/76e/778932/5/check/migrate-ovn-hirsute-wallaby-dvr-snat/76eb0e4/index.html : FAILURE in 10m 51s","accounts_in_message":[],"_revision_number":5},{"id":"b48193ff13bc40ca77a8898da65a7b7dc8dc575a","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-04-12 13:52:19.000000000","message":"Patch Set 5:\n\nfocal-wallaby failed because of https://bugs.launchpad.net/charm-neutron-openvswitch/+bug/1923453","accounts_in_message":[],"_revision_number":5},{"id":"043f86467546affbb2e22c4641fa3300314da87d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-04-16 09:54:14.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9ec967193d24da6c8598343ab8332c3e9a4099f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 10:04:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3148fd9d3fdc45febc60e62b1650f7eb : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6926e639a1294e829893a2fb176e9b1c : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/411d7a54cf2c4ee88694b530c3d2bbf4 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abea164c1dfb416987e1c89763c690f2 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/069d3b8822d24afdb6bdfcf0548b22e2 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/507f8e6dfce24deaad6b8476d76a5a5f : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":6},{"id":"6f9f47251ba0fec617e2646cd472cc3e44a3b834","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-16 12:49:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/6/23485/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/6/23485/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/6/23485/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/6/23485/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":6},{"id":"39cc56af7477a79a2d8d591de75412e590a10d61","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2021-04-20 18:23:35.000000000","message":"Patch Set 6:\n\ncharm-recheck-full","accounts_in_message":[],"_revision_number":6},{"id":"8bc0b1faddd51e7d9c19dfc55bf80d08e45e0083","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-20 20:28:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-neutron-openvswitch/778932/6/8466/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-neutron-openvswitch/778932/6/8466/index.html : FAILURE\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_full/openstack/charm-neutron-openvswitch/778932/6/8466/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":6},{"id":"f257a7db18e90bdf5ab5d9d091b6e5288adb466e","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-21 06:04:40.000000000","message":"Patch Set 6:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/4cc/778932/6/check/osci-lint/4cc2fa4/index.html : SUCCESS in 10m 34s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/7ab/778932/6/check/tox-py35/7abfb16/index.html : SUCCESS in 22m 01s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/ee4/778932/6/check/tox-py36/ee44e01/index.html : SUCCESS in 15m 44s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/e09/778932/6/check/tox-py37/e09881a/index.html : SUCCESS in 22m 21s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/d54/778932/6/check/tox-py38/d54a27e/index.html : SUCCESS in 13m 04s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/b61/778932/6/check/xenial-mitaka_ovs/b61472f/index.html : SUCCESS in 52m 05s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/3b3/778932/6/check/bionic-queens-dvr-snat/3b311d6/index.html : SUCCESS in 1h 35m 04s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/437/778932/6/check/bionic-stein-dvr-snat/4372b2a/index.html : SUCCESS in 1h 18m 28s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/7c5/778932/6/check/bionic-train-dvr-snat/7c56749/index.html : SUCCESS in 1h 19m 52s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/fc5/778932/6/check/bionic-ussuri-dvr-snat/fc54a62/index.html : SUCCESS in 1h 15m 44s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ce2/778932/6/check/migrate-ovn-focal-ussuri-dvr-snat/ce2e950/index.html : SUCCESS in 1h 47m 31s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/01e/778932/6/check/migrate-ovn-focal-victoria-dvr-snat/01e2b71/index.html : SUCCESS in 1h 11m 04s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/4d8/778932/6/check/migrate-ovn-focal-wallaby-dvr-snat/4d8bc46/index.html : FAILURE in 42m 05s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/a08/778932/6/check/migrate-ovn-hirsute-wallaby-dvr-snat/a08f2b3/index.html : FAILURE in 9m 53s","accounts_in_message":[],"_revision_number":6},{"id":"749cf606e8514ab044d142ff7952a7053cc43086","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-05-06 10:18:34.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"2345df324b9eb0cc120cba39c35afe78a773f83c","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-05-06 10:18:42.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"8bcd45e94a5d84b808c225d55b0439e395a5ace6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 10:27:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9935434ea744e4fa14e19eeb8e72a14 : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e7cfb59834484830921bf4e3d6d4eb3d : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57de8c92d56b49428052cee0d9a8ff09 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48c7e0bfee0944b2b708d58585e5e05b : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/329bf658c53d4a769cfb3533592c6821 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af8254e2e85a495983f87f72492381c1 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":7},{"id":"3194997cb3fbed842d2afeadefe283236a85a435","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-05-06 13:49:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/7/23873/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/7/23873/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/7/23873/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/7/23873/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":7},{"id":"f77dd36773bc5c0490ffd56b85934597c5a163d6","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-05-07 07:38:05.000000000","message":"Patch Set 7:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/f31/778932/7/check/osci-lint/f31e764/index.html : SUCCESS in 6m 09s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/73b/778932/7/check/tox-py35/73b5c01/index.html : SUCCESS in 7m 32s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/4a7/778932/7/check/tox-py36/4a7471e/index.html : SUCCESS in 7m 08s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/c79/778932/7/check/tox-py37/c796614/index.html : SUCCESS in 8m 32s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/dd2/778932/7/check/tox-py38/dd2dade/index.html : SUCCESS in 7m 28s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/b4d/778932/7/check/xenial-mitaka_ovs/b4dea3d/index.html : SUCCESS in 49m 54s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/96b/778932/7/check/bionic-queens-dvr-snat/96b56a2/index.html : SUCCESS in 1h 05m 32s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/952/778932/7/check/bionic-stein-dvr-snat/95291ca/index.html : SUCCESS in 1h 20m 03s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/202/778932/7/check/bionic-train-dvr-snat/20254fb/index.html : FAILURE in 26m 38s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/38c/778932/7/check/bionic-ussuri-dvr-snat/38c2205/index.html : SUCCESS in 1h 14m 17s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/427/778932/7/check/migrate-ovn-focal-ussuri-dvr-snat/427af2a/index.html : SUCCESS in 1h 17m 03s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/a1f/778932/7/check/migrate-ovn-focal-victoria-dvr-snat/a1f2be0/index.html : SUCCESS in 1h 12m 08s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/215/778932/7/check/migrate-ovn-focal-wallaby-dvr-snat/21535bd/index.html : TIMED_OUT in 2h 17m 42s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/267/778932/7/check/migrate-ovn-hirsute-wallaby-dvr-snat/2679409/index.html : FAILURE in 9m 54s","accounts_in_message":[],"_revision_number":7},{"id":"795e5d4fb8572043cf171798057a351d8e7dfb60","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-05-07 08:26:51.000000000","message":"Patch Set 7:\n\nNeutronNetworkingTest failed on focal-wallaby at pinging a new floating IP. This may be a legit problem in the product. Re-checking","accounts_in_message":[],"_revision_number":7},{"id":"55b1b7f05748a9c58e80b0ff0151a3f3e947bcf1","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-05-07 08:26:57.000000000","message":"Patch Set 7:\n\nzosci-recheck","accounts_in_message":[],"_revision_number":7},{"id":"f60647cdedd449e5fa62cdf06988530817c525f9","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-05-07 09:30:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/7/23898/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/7/23898/index.html : FAILURE\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/778932/7/23898/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":7},{"id":"7c4efc9f9047b34d354c0d3336377acdfef5d9e1","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-05-07 20:12:47.000000000","message":"Patch Set 7:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/753/778932/7/check/osci-lint/75358aa/index.html : SUCCESS in 5m 24s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/160/778932/7/check/tox-py35/1602100/index.html : SUCCESS in 7m 35s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/e7e/778932/7/check/tox-py36/e7ef638/index.html : SUCCESS in 6m 05s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/d42/778932/7/check/tox-py37/d42c3e1/index.html : SUCCESS in 7m 46s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/f9f/778932/7/check/tox-py38/f9fa4e1/index.html : SUCCESS in 6m 07s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/73f/778932/7/check/xenial-mitaka_ovs/73fd817/index.html : TIMED_OUT in 2h 06m 58s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/cf6/778932/7/check/bionic-queens-dvr-snat/cf69f4a/index.html : FAILURE in 32m 13s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/fda/778932/7/check/bionic-stein-dvr-snat/fdae5b4/index.html : SUCCESS in 1h 32m 35s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/47e/778932/7/check/bionic-train-dvr-snat/47e3914/index.html : SUCCESS in 1h 27m 42s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/475/778932/7/check/bionic-ussuri-dvr-snat/475a9b3/index.html : SUCCESS in 1h 37m 13s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/217/778932/7/check/migrate-ovn-focal-ussuri-dvr-snat/217bc3e/index.html : SUCCESS in 1h 23m 06s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/4bc/778932/7/check/migrate-ovn-focal-victoria-dvr-snat/4bcbaaa/index.html : SUCCESS in 1h 29m 56s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/854/778932/7/check/migrate-ovn-focal-wallaby-dvr-snat/854bdd9/index.html : TIMED_OUT in 2h 17m 40s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/811/778932/7/check/migrate-ovn-hirsute-wallaby-dvr-snat/81191fc/index.html : FAILURE in 9m 48s","accounts_in_message":[],"_revision_number":7},{"id":"23b8a49d025e1735b2012ea8eadc0635adb69adc","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-05-07 20:20:16.000000000","message":"Patch Set 7: Code-Review-1\n\nUnfortunately confirmed: it\u0027s the second time that NeutronNetworkingTest failed on focal-wallaby at pinging a new floating IP","accounts_in_message":[],"_revision_number":7},{"id":"b2b4127e719bc47b6db1957e2242d100e7da8ba6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-06-15 12:54:34.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"95d7a8f302bc19d7207ddf0a596004ee9dcf38a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 13:08:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b91527e64e5442999d6d8688d86f7f37 : SUCCESS in 5m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3310ba678a564333a7010b82505f8a59 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06ca2f28ab2247829e7f61d402de8b32 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec08711a92c5469894c70da904a90578 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a08fa548cdb343b88e73cea34e07b1ff : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b743058de77a4749b8280cc5f9b258fa : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":8},{"id":"d0fd952ae5da28e40507f3c7665d907dd2257866","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-06-15 21:36:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/883/778932/8/check/osci-lint/883ef41/index.html : SUCCESS in 7m 42s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/726/778932/8/check/tox-py35/7261d43/index.html : SUCCESS in 7m 54s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/a8c/778932/8/check/tox-py36/a8c897a/index.html : SUCCESS in 8m 01s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/b53/778932/8/check/tox-py37/b53290f/index.html : SUCCESS in 7m 16s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/e33/778932/8/check/tox-py38/e33a6d7/index.html : SUCCESS in 8m 56s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/5c2/778932/8/check/xenial-mitaka_ovs/5c2048e/index.html : FAILURE in 25m 50s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/e9b/778932/8/check/bionic-queens-dvr-snat/e9b3802/index.html : SUCCESS in 1h 48m 47s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/1aa/778932/8/check/bionic-stein-dvr-snat/1aae126/index.html : SUCCESS in 1h 41m 46s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/0b0/778932/8/check/bionic-train-dvr-snat/0b0674a/index.html : FAILURE in 27m 03s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/be3/778932/8/check/bionic-ussuri-dvr-snat/be3f1e7/index.html : FAILURE in 42m 48s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/8e8/778932/8/check/migrate-ovn-focal-ussuri-dvr-snat/8e8c815/index.html : SUCCESS in 1h 24m 17s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d0a/778932/8/check/migrate-ovn-focal-victoria-dvr-snat/d0ab88b/index.html : FAILURE in 26m 49s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/b7a/778932/8/check/migrate-ovn-focal-wallaby-dvr-snat/b7a95be/index.html : FAILURE in 15m 03s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/6d8/778932/8/check/migrate-ovn-hirsute-wallaby-dvr-snat/6d868bf/index.html : FAILURE in 31m 18s","accounts_in_message":[],"_revision_number":8},{"id":"48b51257d977f357f4330909ea84125b781f3fc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-06-28 13:21:43.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a48bf32409cca1d1ce427907fda76fbf65000f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:31:30.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e83fba61f604246a026e3f3d2195beb : SUCCESS in 7m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/96f87b0272bb465396b7352a06c40e7f : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c97d699c12a4f758fa1170227de62aa : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df3bce063a534649a95c058b4a954052 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13b569f1a19b477092f00f1f1facfd57 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fab02f245bb405fb034a97a85ca1d35 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":9},{"id":"2f1fac126aa738b91abfa25417d17ae1cc185f31","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-06-28 18:09:39.000000000","message":"Patch Set 9: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/3b6/778932/9/check/osci-lint/3b6e55a/index.html : SUCCESS in 6m 44s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/300/778932/9/check/tox-py35/300d0b8/index.html : SUCCESS in 5m 16s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/125/778932/9/check/tox-py36/1251a52/index.html : SUCCESS in 5m 36s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/34e/778932/9/check/tox-py37/34ef13d/index.html : SUCCESS in 6m 02s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/986/778932/9/check/tox-py38/986f286/index.html : SUCCESS in 5m 42s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/74f/778932/9/check/migrate-ovn-focal-ussuri-dvr-snat/74f1ded/index.html : FAILURE in 1h 28m 10s\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED\n- migrate-ovn-focal-wallaby-dvr-snat migrate-ovn-focal-wallaby-dvr-snat : SKIPPED\n- migrate-ovn-groovy-victoria-dvr-snat migrate-ovn-groovy-victoria-dvr-snat : SKIPPED\n- migrate-ovn-hirsute-wallaby-dvr-snat migrate-ovn-hirsute-wallaby-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":9},{"id":"a0e9e94465ca6691e5d0abe569c414407fc6ca44","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-06-29 09:00:21.000000000","message":"Patch Set 9:\n\nThis looks like a deployment timeout, re-triggering","accounts_in_message":[],"_revision_number":9},{"id":"b39c0e1214e9e1888d4ad017ded01e909a36ba82","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-06-29 09:00:33.000000000","message":"Patch Set 9:\n\ncharm-recheck","accounts_in_message":[],"_revision_number":9},{"id":"c4c8027656ac69651b3ad1b12911d0b16fafc06c","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-06-29 10:35:14.000000000","message":"Patch Set 9:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/7d1/778932/9/check/osci-lint/7d102e9/index.html : SUCCESS in 5m 39s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/f63/778932/9/check/tox-py35/f633d12/index.html : SUCCESS in 5m 59s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/db0/778932/9/check/tox-py36/db00e9f/index.html : SUCCESS in 6m 01s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/a28/778932/9/check/tox-py37/a282212/index.html : SUCCESS in 6m 04s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/ef6/778932/9/check/tox-py38/ef6fbd0/index.html : SUCCESS in 6m 12s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/5f5/778932/9/check/migrate-ovn-focal-ussuri-dvr-snat/5f56267/index.html : FAILURE in 1h 27m 32s\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED\n- migrate-ovn-focal-wallaby-dvr-snat migrate-ovn-focal-wallaby-dvr-snat : SKIPPED\n- migrate-ovn-groovy-victoria-dvr-snat migrate-ovn-groovy-victoria-dvr-snat : SKIPPED\n- migrate-ovn-hirsute-wallaby-dvr-snat migrate-ovn-hirsute-wallaby-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":9},{"id":"371d80ed60e07fef2efc020f8cb4c61a9e6482fb","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-06-29 11:57:01.000000000","message":"Patch Set 9:\n\nI think Zaza\u0027s model readiness detection is broken since https://github.com/openstack-charmers/zaza/pull/451 landed","accounts_in_message":[],"_revision_number":9},{"id":"67701e467a21d52bfe081106ca21b21eab326fcc","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-06-29 13:08:10.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9:\n\u003e \n\u003e I think Zaza\u0027s model readiness detection is broken since https://github.com/openstack-charmers/zaza/pull/451 landed\n\nUnfortunately, zaza\u0027s readiness detection has improved and now (correctly) detects the neutron-api-plugin-ovn application as \"not ready\" as it has no units.  This can be resolved in the bundle by setting \"num-expected-units: 0\" for the application.  Unfortunately, this then breaks the tests as the tests assume that just adding the units will work and continue to use the same \u0027readiness\u0027 criteria from the bundle.  Thus:\n\n1. The bundle needs to be updated with num-expected-units: 0 for the neutron-api-plugin-ovn application.\n2. The tests in zaza-openstack-tests need to be updated to check for the right criteria when the topology is changed (neutron-api-plugin-ovn is related).\n\nI\u0027ll do that as a separate set of changes.","accounts_in_message":[],"_revision_number":9},{"id":"da025abe5b86e3bcf465dae5e0875d6fe41794e5","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-07-05 09:28:05.000000000","message":"Patch Set 9:\n\nhttps://review.opendev.org/c/openstack/charm-neutron-gateway/+/799443","accounts_in_message":[],"_revision_number":9},{"id":"892400b9f8f501452c249ddf0606d5c237a7dfa5","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-07-06 08:00:41.000000000","message":"Patch Set 9:\n\nhttps://review.opendev.org/c/openstack/charm-neutron-openvswitch/+/799570","accounts_in_message":[],"_revision_number":9},{"id":"544bb6c812c7e88767c6a8c38e1fa2e520d4ff6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-07-06 12:11:18.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"28dcd165e374d7ac1d7b62fc260e4895d6e1c695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 12:21:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3429fbc93c6a4577a2615c3ff76b350d : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f3808fe733624ac9a1611d714515cc45 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ed538668f724e938e403abe27a2bde6 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26248a66730b4315bbf314893ea49e0d : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd34a8880c4942bca3b95a0a7d50c1d3 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53728d4b96ad400e9abe329331b5b358 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":10},{"id":"b3ab3d2206232d8b43f79eb33396b69c60f9f304","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-07-06 17:56:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/1c6/778932/10/check/osci-lint/1c66913/index.html : SUCCESS in 4m 55s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/00b/778932/10/check/tox-py35/00be927/index.html : SUCCESS in 6m 01s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/6af/778932/10/check/tox-py36/6af6713/index.html : SUCCESS in 5m 55s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/2db/778932/10/check/tox-py37/2dbb9c6/index.html : SUCCESS in 5m 54s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/ca3/778932/10/check/tox-py38/ca37591/index.html : SUCCESS in 5m 33s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/1e7/778932/10/check/xenial-mitaka_ovs/1e7635a/index.html : SUCCESS in 57m 25s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/4de/778932/10/check/bionic-queens-dvr-snat/4def312/index.html : SUCCESS in 1h 07m 52s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/e6d/778932/10/check/bionic-stein-dvr-snat/e6d2383/index.html : SUCCESS in 1h 37m 54s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d02/778932/10/check/bionic-train-dvr-snat/d0287a1/index.html : SUCCESS in 1h 38m 12s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/eac/778932/10/check/bionic-ussuri-dvr-snat/eac844a/index.html : SUCCESS in 1h 29m 08s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/129/778932/10/check/migrate-ovn-focal-ussuri-dvr-snat/129d83e/index.html : SUCCESS in 1h 16m 13s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/878/778932/10/check/migrate-ovn-focal-victoria-dvr-snat/878fbf0/index.html : SUCCESS in 1h 27m 10s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/e69/778932/10/check/migrate-ovn-focal-wallaby-dvr-snat/e690806/index.html : TIMED_OUT in 3h 15m 43s\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/7db/778932/10/check/migrate-ovn-groovy-victoria-dvr-snat/7db746c/index.html : SUCCESS in 1h 11m 37s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/3d3/778932/10/check/migrate-ovn-hirsute-wallaby-dvr-snat/3d30a36/index.html : TIMED_OUT in 3h 13m 49s","accounts_in_message":[],"_revision_number":10},{"id":"cdac36570b0d3490fc97e09830bc193d2c66ec07","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-07-07 06:38:26.000000000","message":"Patch Set 10:\n\nBoth wallaby-dvr-snat bundles timed out in the same way. Strange. Re-triggering","accounts_in_message":[],"_revision_number":10},{"id":"4f8fb2c3ac8d6327017bc4ef29ceddd45b739c21","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-07-07 06:40:38.000000000","message":"Patch Set 10:\n\ncharm-recheck","accounts_in_message":[],"_revision_number":10},{"id":"1fdd2436c72421870812b1d47a630aac192e56a2","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-07-07 11:04:10.000000000","message":"Patch Set 10:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/def/778932/10/check/osci-lint/def61f8/index.html : SUCCESS in 5m 54s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/8dd/778932/10/check/tox-py35/8dd8c0b/index.html : SUCCESS in 7m 40s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/431/778932/10/check/tox-py36/4318913/index.html : SUCCESS in 6m 42s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/b6b/778932/10/check/tox-py37/b6b7ebe/index.html : SUCCESS in 7m 06s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/0ed/778932/10/check/tox-py38/0edd955/index.html : SUCCESS in 7m 10s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/80c/778932/10/check/xenial-mitaka_ovs/80c3a9b/index.html : TIMED_OUT in 3h 06m 17s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/c98/778932/10/check/bionic-queens-dvr-snat/c988bfc/index.html : FAILURE in 16m 32s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/5ef/778932/10/check/bionic-stein-dvr-snat/5ef7176/index.html : SUCCESS in 1h 18m 05s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/2e2/778932/10/check/bionic-train-dvr-snat/2e219b4/index.html : SUCCESS in 1h 11m 43s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/5f3/778932/10/check/bionic-ussuri-dvr-snat/5f3cd14/index.html : FAILURE in 14m 34s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/ae9/778932/10/check/migrate-ovn-focal-ussuri-dvr-snat/ae9303b/index.html : SUCCESS in 1h 08m 09s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/bcc/778932/10/check/migrate-ovn-focal-victoria-dvr-snat/bccefc7/index.html : FAILURE in 13m 44s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/01f/778932/10/check/migrate-ovn-focal-wallaby-dvr-snat/01fe0e5/index.html : FAILURE in 13m 29s\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/984/778932/10/check/migrate-ovn-groovy-victoria-dvr-snat/984e7b2/index.html : FAILURE in 13m 04s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/7eb/778932/10/check/migrate-ovn-hirsute-wallaby-dvr-snat/7ebabb3/index.html : FAILURE in 13m 30s","accounts_in_message":[],"_revision_number":10},{"id":"33e27cf34e4311bbc8325a436463aafcf2f782e8","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-07-08 11:06:38.000000000","message":"Patch Set 10:\n\nInfra issues while accessing the charmstore. Re-triggering","accounts_in_message":[],"_revision_number":10},{"id":"4f58920465753b47f2e69889fca687c85120b088","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-07-08 11:06:48.000000000","message":"Patch Set 10:\n\ncharm-recheck","accounts_in_message":[],"_revision_number":10},{"id":"5d782cda9143c19f298525650690f3f85940973d","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-07-08 15:30:05.000000000","message":"Patch Set 10:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/286/778932/10/check/osci-lint/286f124/index.html : SUCCESS in 4m 40s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/8e3/778932/10/check/tox-py35/8e3da28/index.html : SUCCESS in 5m 10s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/a35/778932/10/check/tox-py36/a359e4b/index.html : SUCCESS in 5m 45s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/04d/778932/10/check/tox-py37/04dd7a1/index.html : SUCCESS in 5m 18s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/349/778932/10/check/tox-py38/3496c25/index.html : SUCCESS in 6m 02s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/14f/778932/10/check/xenial-mitaka_ovs/14fab8d/index.html : SUCCESS in 50m 12s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/138/778932/10/check/bionic-queens-dvr-snat/138caba/index.html : SUCCESS in 1h 14m 08s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/096/778932/10/check/bionic-stein-dvr-snat/0968294/index.html : SUCCESS in 1h 13m 36s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/78c/778932/10/check/bionic-train-dvr-snat/78cd00b/index.html : SUCCESS in 1h 21m 50s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/88c/778932/10/check/bionic-ussuri-dvr-snat/88c4a29/index.html : SUCCESS in 1h 20m 00s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/169/778932/10/check/migrate-ovn-focal-ussuri-dvr-snat/169be71/index.html : SUCCESS in 1h 00m 16s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/772/778932/10/check/migrate-ovn-focal-victoria-dvr-snat/77278c3/index.html : SUCCESS in 1h 10m 30s\n- migrate-ovn-focal-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/21a/778932/10/check/migrate-ovn-focal-wallaby-dvr-snat/21a7de6/index.html : TIMED_OUT in 3h 14m 16s\n- migrate-ovn-groovy-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/278/778932/10/check/migrate-ovn-groovy-victoria-dvr-snat/2784c4e/index.html : SUCCESS in 1h 16m 29s\n- migrate-ovn-hirsute-wallaby-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/f8c/778932/10/check/migrate-ovn-hirsute-wallaby-dvr-snat/f8c500b/index.html : TIMED_OUT in 3h 15m 27s","accounts_in_message":[],"_revision_number":10},{"id":"1b8033b30cd875dd4c5b449a570c7927fe353fdc","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-07-12 11:01:22.000000000","message":"Patch Set 10: Code-Review-1\n\nhttps://bugs.launchpad.net/charm-neutron-openvswitch/+bug/1935828","accounts_in_message":[],"_revision_number":10},{"id":"8f4946ec816f3801812427e473a7585c1f146768","tag":"autogenerated:gerrit:abandon","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2021-10-05 18:03:26.000000000","message":"Abandoned\n\nAbandoning in favor of xena-bundle that includes the focal-w and hirsute-w bundles in this patchset: I26c02161130e1715d80cc8dbb0bcece77d2df249","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"470ae515fc47a515d98715c051fe5b17ddb1731b","revisions":{"bb84b6feff872523bba2d9ca4fded625d478888f":{"kind":"REWORK","_number":1,"created":"2021-03-05 14:11:49.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/32/778932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/1"}}},"commit":{"parents":[{"commit":"3722e34245594a3db6fecbddeb3737ff6598ae23","subject":"Add hirsute and remove trusty from metadata.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/3722e34245594a3db6fecbddeb3737ff6598ae23"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/bb84b6feff872523bba2d9ca4fded625d478888f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/bb84b6feff872523bba2d9ca4fded625d478888f"}]},"branch":"refs/heads/master"},"be53667b10ab2311db4f6b965eeb3e8f5416c081":{"kind":"REWORK","_number":2,"created":"2021-03-15 12:45:37.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/32/778932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/2"}}},"commit":{"parents":[{"commit":"2699a79b95a96699c16b130fc049e42dec873238","subject":"Add hirsute and remove trusty from metadata.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/2699a79b95a96699c16b130fc049e42dec873238"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-15 12:45:18.000000000","tz":0},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/be53667b10ab2311db4f6b965eeb3e8f5416c081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/be53667b10ab2311db4f6b965eeb3e8f5416c081"}]},"branch":"refs/heads/master"},"10b0e81aedef62331d64e01b069c4506836ccde1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-23 13:15:58.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/32/778932/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/3"}}},"commit":{"parents":[{"commit":"a2a3c6957e0ed522040dd74632566b84ef9b2dde","subject":"Add hirsute and remove trusty from metadata.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/a2a3c6957e0ed522040dd74632566b84ef9b2dde"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-23 13:14:20.000000000","tz":0},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/10b0e81aedef62331d64e01b069c4506836ccde1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/10b0e81aedef62331d64e01b069c4506836ccde1"}]},"branch":"refs/heads/master"},"25a6bf6b5d5304a4ff3ee5fa7484a8c3f5adaeb7":{"kind":"REWORK","_number":4,"created":"2021-03-30 17:29:49.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/32/778932/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/4"}}},"commit":{"parents":[{"commit":"dfbe69a62a7cd7f2b464c9c50a39f46b349b1a82","subject":"Add hirsute and remove trusty from metadata.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/dfbe69a62a7cd7f2b464c9c50a39f46b349b1a82"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-30 17:28:26.000000000","tz":60},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nLibraries sync.\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/25a6bf6b5d5304a4ff3ee5fa7484a8c3f5adaeb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/25a6bf6b5d5304a4ff3ee5fa7484a8c3f5adaeb7"}]},"branch":"refs/heads/master"},"314738cf78c66e59da401e4d685e6193f010e17b":{"kind":"REWORK","_number":5,"created":"2021-04-10 15:09:36.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/32/778932/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/5"}}},"commit":{"parents":[{"commit":"9b11d24d40d4c8dfffe90a264e0273cacfd676f3","subject":"Implementation of deferred restarts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/9b11d24d40d4c8dfffe90a264e0273cacfd676f3"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","date":"2021-04-10 15:08:49.000000000","tz":60},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nLibraries sync.\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/314738cf78c66e59da401e4d685e6193f010e17b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/314738cf78c66e59da401e4d685e6193f010e17b"}]},"branch":"refs/heads/master"},"f0688d21646843e8603e10ddbfda22a964fb93fc":{"kind":"REWORK","_number":6,"created":"2021-04-16 09:54:14.000000000","uploader":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"ref":"refs/changes/32/778932/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/6"}}},"commit":{"parents":[{"commit":"0653da1aa6e1c8dad57ad701431d53a7c2eadeb5","subject":"21.04 sync - add \u0027hirsute\u0027 in UBUNTU_RELEASES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/0653da1aa6e1c8dad57ad701431d53a7c2eadeb5"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-04-16 09:54:12.000000000","tz":60},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nLibraries sync.\n\nhisute-wallaby test is disabled (moved to dev) due to [1] as bundle may\nreference a reactive charm.\n\n[1] https://github.com/juju-solutions/layer-basic/issues/194\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/f0688d21646843e8603e10ddbfda22a964fb93fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/f0688d21646843e8603e10ddbfda22a964fb93fc"}]},"branch":"refs/heads/master"},"90e4d414ff82c27dc7a7528b3af893c00458be9a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-05-06 10:18:34.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/32/778932/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/7"}}},"commit":{"parents":[{"commit":"561e22d2629e6e597b38b30ec783a774960be41a","subject":"Updates to flip all libraries back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/561e22d2629e6e597b38b30ec783a774960be41a"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-05-06 10:18:34.000000000","tz":0},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nLibraries sync.\n\nhisute-wallaby test is disabled (moved to dev) due to [1] as bundle may\nreference a reactive charm.\n\n[1] https://github.com/juju-solutions/layer-basic/issues/194\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/90e4d414ff82c27dc7a7528b3af893c00458be9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/90e4d414ff82c27dc7a7528b3af893c00458be9a"}]},"branch":"refs/heads/master","description":"Rebase"},"171874f64168d9829c020aee1bbbaea89ed06930":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-06-15 12:54:34.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/32/778932/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/8"}}},"commit":{"parents":[{"commit":"6ad570051cd67d1946b1246e136d4a1a245b2a5e","subject":"Merge \"c-h sync - restore proxy env vars for add-apt-repository\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/6ad570051cd67d1946b1246e136d4a1a245b2a5e"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-06-15 12:54:34.000000000","tz":0},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nLibraries sync.\n\nhisute-wallaby test is disabled (moved to dev) due to [1] as bundle may\nreference a reactive charm.\n\n[1] https://github.com/juju-solutions/layer-basic/issues/194\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/171874f64168d9829c020aee1bbbaea89ed06930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/171874f64168d9829c020aee1bbbaea89ed06930"}]},"branch":"refs/heads/master","description":"Rebase"},"31d0b61175b19fcf23a5b144d3f9649ea60b0b2c":{"kind":"REWORK","_number":9,"created":"2021-06-28 13:21:43.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/32/778932/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/9"}}},"commit":{"parents":[{"commit":"3fc68108afbd3ce5caf46ea3e5eae05a920f6c0c","subject":"Merge \"Deprecate linux bridge usage in data-port config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/3fc68108afbd3ce5caf46ea3e5eae05a920f6c0c"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-06-28 13:21:35.000000000","tz":120},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nLibraries sync.\n\nhisute-wallaby test is disabled (moved to dev) due to [1] as bundle may\nreference a reactive charm.\n\n[1] https://github.com/juju-solutions/layer-basic/issues/194\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\nCo-authored-by: Aurelien Lourot \u003caurelien.lourot@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/31d0b61175b19fcf23a5b144d3f9649ea60b0b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/31d0b61175b19fcf23a5b144d3f9649ea60b0b2c"}]},"branch":"refs/heads/master"},"470ae515fc47a515d98715c051fe5b17ddb1731b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-07-06 12:11:18.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/32/778932/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/32/778932/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/32/778932/10"}}},"commit":{"parents":[{"commit":"7ad1afeb741d3273ef6e82fdbd5b263483396c99","subject":"Fix Zaza deployment expectations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/7ad1afeb741d3273ef6e82fdbd5b263483396c99"}]}],"author":{"name":"Alex Kavanagh","email":"alex.kavanagh@canonical.com","date":"2021-03-05 14:11:35.000000000","tz":0},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2021-07-06 12:11:11.000000000","tz":120},"subject":"Test bundles for focal-wallaby and hirsute-wallaby","message":"Test bundles for focal-wallaby and hirsute-wallaby\n\nThese are the test bundles (and any associated changes) for\nfocal-wallaby and hirsute-wallaby support.\n\nLibraries sync.\n\nhisute-wallaby test is disabled (moved to dev) due to [1] as bundle may\nreference a reactive charm.\n\n[1] https://github.com/juju-solutions/layer-basic/issues/194\n\nChange-Id: I07a7c57f55822d3fa0e24133cf6e166d12945ebe\nCo-authored-by: Aurelien Lourot \u003caurelien.lourot@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/470ae515fc47a515d98715c051fe5b17ddb1731b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/470ae515fc47a515d98715c051fe5b17ddb1731b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
