)]}'
{"id":"openstack%2Fcharm-neutron-openvswitch~787448","triplet_id":"openstack%2Fcharm-neutron-openvswitch~master~I62c66648898ba3fe59ec8523542b7a82a8b6c886","project":"openstack/charm-neutron-openvswitch","branch":"master","topic":"lp1912847","attention_set":{},"removed_from_attention_set":{"25011":{"account":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"last_update":"2022-09-12 10:52:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I62c66648898ba3fe59ec8523542b7a82a8b6c886","subject":"Add NRPE checks for openvswitch","status":"ABANDONED","created":"2021-04-21 21:54:45.000000000","updated":"2022-09-12 10:52:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ccf6d34f3f89bb1c4e0cd81e57f95e8e7fea0e3d","_number":787448,"virtual_id_number":787448,"owner":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},{"value":-1,"date":"2021-04-26 16:12:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-26 16:21:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":30028,"name":"James Troup","email":"ubuntu@elmo.tasta.io","username":"elmo"}],"REVIEWER":[{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-21 22:17:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 00:33:25.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-22 00:52:37.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-08 22:07:32.000000000","updated_by":{"_account_id":30028,"name":"James Troup","email":"ubuntu@elmo.tasta.io","username":"elmo"},"reviewer":{"_account_id":30028,"name":"James Troup","email":"ubuntu@elmo.tasta.io","username":"elmo"},"state":"CC"},{"updated":"2022-09-12 10:52:50.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"CC"}],"messages":[{"id":"b1198826b9b0278c1b9c52403106695208b72963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"date":"2021-04-21 21:54:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c1c378d0ba916247f672f3365691bdf03974dbc","author":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"date":"2021-04-21 21:55:36.000000000","message":"Patch Set 1: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":1},{"id":"54c43748fd3f3fef1573b9384ba9cd12648e0a4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 22:17:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae469b88123c41a2beeaa141e87659ef : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0c4814e5b6e24b378f9e470fc82c15d7 : FAILURE in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b67db4cc4764cc3b39fc347f170f635 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69e42990b47a4d7aa982e8f982a7a927 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42d674ad66f04c8ba40377cedd987629 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49b8396bcfc94685a5e2f58c6d589cbb : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a47f189698295ec5f72d7c3aaada6e73a90132c5","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-22 00:33:25.000000000","message":"Patch Set 1:\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/5e3/787448/1/check/osci-lint/5e31ea0/index.html : SUCCESS in 5m 12s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/fa4/787448/1/check/tox-py35/fa42aa1/index.html : SUCCESS in 6m 16s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/112/787448/1/check/tox-py36/11286a2/index.html : SUCCESS in 5m 23s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/e26/787448/1/check/tox-py37/e264104/index.html : SUCCESS in 5m 40s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/34d/787448/1/check/tox-py38/34d8fd9/index.html : SUCCESS in 5m 39s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/6b0/787448/1/check/xenial-mitaka_ovs/6b08d52/index.html : SUCCESS in 55m 41s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/351/787448/1/check/bionic-queens-dvr-snat/3515f49/index.html : SUCCESS in 1h 08m 54s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/cb0/787448/1/check/bionic-stein-dvr-snat/cb0ccac/index.html : SUCCESS in 1h 20m 49s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/9a0/787448/1/check/bionic-train-dvr-snat/9a01e6a/index.html : SUCCESS in 1h 28m 54s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/f63/787448/1/check/bionic-ussuri-dvr-snat/f635ea0/index.html : SUCCESS in 1h 15m 01s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/47e/787448/1/check/migrate-ovn-focal-ussuri-dvr-snat/47effe5/index.html : SUCCESS in 1h 02m 15s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/836/787448/1/check/migrate-ovn-focal-victoria-dvr-snat/8368927/index.html : SUCCESS in 1h 13m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1ed283d4a7b0a092d72391a1864479055e0ca4bc","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-22 00:52:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/1/23556/index.html : SUCCESS\n* test_charm_func_smoke https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/1/23556/index.html : SUCCESS\n* test_charm_single https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/1/23556/index.html : SUCCESS\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/1/23556/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":1},{"id":"6217913b59107227a3ba2beae449555d1c628b38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"date":"2021-04-23 15:09:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6a87e60c5f2956dce86f43bf34e6eb505c70bcf2","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-23 15:11:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/2/23582/index.html : FAILURE\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/2/23582/index.html : FAILURE\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":2},{"id":"db8f80255ebefc0375974858adb9bb4efa3330e6","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-23 15:18:28.000000000","message":"Patch Set 2:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/d63/787448/2/check/osci-lint/d634527/index.html : FAILURE in 5m 04s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/7ba/787448/2/check/tox-py35/7ba9798/index.html : FAILURE in 5m 54s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/e6c/787448/2/check/tox-py36/e6c6f41/index.html : FAILURE in 5m 04s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/4b1/787448/2/check/tox-py37/4b1cde1/index.html : FAILURE in 6m 18s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/217/787448/2/check/tox-py38/2179fd3/index.html : FAILURE in 5m 21s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat migrate-ovn-focal-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"2db5d34df8eafdfdcc6868e285fe189240e65be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 17:00:02.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50df4f88cc9d47d2824034f67b3567f2 : FAILURE in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/262f2ec38247498282d0c797fb920d0a : FAILURE in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27d197e296994d579a39f19e85a5c8a4 : FAILURE in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49b8b6ced1d5443cabcf1b012c00b5ef : FAILURE in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c55f1390cbd7431c848c9cad3c99a8f6 : FAILURE in 5m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d71f7b758b0457fa81a829e98477835 : FAILURE in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f4dd0b25fce01995c50584e9c4f800f641950b27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"date":"2021-04-23 19:05:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d746d67e2dc9cc4ae78508432eacb5b43d84478f","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-23 19:07:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/3/23586/index.html : FAILURE\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/3/23586/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":3},{"id":"c88869878f94fe5e537c36263d10a9018bb28b78","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-23 19:13:23.000000000","message":"Patch Set 3:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/b5a/787448/3/check/osci-lint/b5adb56/index.html : SUCCESS in 6m 56s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/b83/787448/3/check/tox-py35/b83250e/index.html : FAILURE in 5m 06s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/64c/787448/3/check/tox-py36/64ccf2f/index.html : FAILURE in 5m 05s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/a59/787448/3/check/tox-py37/a59c667/index.html : FAILURE in 6m 32s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/1ac/787448/3/check/tox-py38/1ac0d46/index.html : FAILURE in 6m 58s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat migrate-ovn-focal-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":3},{"id":"32ec267d175d1cd9fb1f0a38aa4425737f4d5454","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 19:50:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9743e9b1e745a4b2a22528c7baaaa5 : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8c790c9fa6eb40e3864cd9d269b8268c : FAILURE in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74f3f86bfdb8457f93500e9bf885874f : FAILURE in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a8ef929cc74444c88c2605cd8b06311 : FAILURE in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c77dbf623c3a48109b5f1bc62659e18f : FAILURE in 5m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05286740d9c44d71be83e5182b787905 : FAILURE in 4m 53s","accounts_in_message":[],"_revision_number":3},{"id":"bb085d3484afec691931800ea99547d2a3d30d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"date":"2021-04-23 21:13:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2c0b1dd81f800517be4056c966ada6ca758f9e99","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-23 21:15:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/4/23590/index.html : FAILURE\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/4/23590/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":4},{"id":"0fe1c2653c00a29f6c58aec810d87c96c579f7d1","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-23 21:22:37.000000000","message":"Patch Set 4:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/729/787448/4/check/osci-lint/7291fe0/index.html : SUCCESS in 8m 32s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/b95/787448/4/check/tox-py35/b95edb5/index.html : FAILURE in 5m 43s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/ba3/787448/4/check/tox-py36/ba3ab27/index.html : FAILURE in 6m 16s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/2b2/787448/4/check/tox-py37/2b2e576/index.html : FAILURE in 6m 51s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/713/787448/4/check/tox-py38/7131b93/index.html : FAILURE in 6m 24s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat migrate-ovn-focal-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":4},{"id":"a9c3e32d05fb061458e3d56b67df92c291752dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 21:23:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bccd15df35124d8d817332918e47f6f7 : SUCCESS in 5m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6068a1eac202447b96563d9217134d4b : FAILURE in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74755b824ddb4161b25a6c960a0c515e : FAILURE in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f98b64aa6c042998b87615a489efd50 : FAILURE in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d0071b659504f7588ca588e0419800b : FAILURE in 5m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7da86875230442448860d0de8e05fdb6 : FAILURE in 5m 30s","accounts_in_message":[],"_revision_number":4},{"id":"e70d3adc9c57ae0a71537801213f44be54e13c36","author":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"date":"2021-04-23 21:45:56.000000000","message":"Patch Set 4: Workflow-1\n\nTests are pending upstream merges.  See layered commit that includes test fixes: https://review.opendev.org/c/openstack/charm-neutron-openvswitch/+/787787","accounts_in_message":[],"_revision_number":4},{"id":"abdae85ad6212bf45b90243d0c3fdc87fd41e6f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"date":"2021-04-26 16:10:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"efd4211e0b4bcc2e6129edd2d47c1b1bbedffe4e","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-26 16:12:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n* test_charm_unit https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/5/23613/index.html : FAILURE\n* test_charm_lint https://openstack-ci-reports.ubuntu.com/artifacts/test_charm_pipeline_func_smoke/openstack/charm-neutron-openvswitch/787448/5/23613/index.html : SUCCESS\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set)\n- \u0027charm-recheck-full\u0027 (run/re-run the full charm functional test set - may take hours!)\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.","accounts_in_message":[],"_revision_number":5},{"id":"1944f3171c55e8bcf98f955df25a83e2843107b5","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-04-26 16:17:32.000000000","message":"Patch Set 5:\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/277/787448/5/check/osci-lint/277fb12/index.html : SUCCESS in 6m 16s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/6c8/787448/5/check/tox-py35/6c8ab6b/index.html : FAILURE in 5m 24s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/e56/787448/5/check/tox-py36/e56ad74/index.html : FAILURE in 6m 35s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/59e/787448/5/check/tox-py37/59edfd1/index.html : FAILURE in 6m 04s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/993/787448/5/check/tox-py38/993b55c/index.html : FAILURE in 6m 34s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat migrate-ovn-focal-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":5},{"id":"00a4bcca0f9aa9a58830ddd0dc06dc0417bcea5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:21:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/945218c7647040cd939ba9d7f06305d4 : SUCCESS in 5m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bc4842b7d8a24ab88441c9849860dd43 : FAILURE in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6785b0908fcb4ffa869ccc5d3ca6ae4a : FAILURE in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2f6bde730984b708776373a8669be74 : FAILURE in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/226bef298c1141eda85af2722c70e71f : FAILURE in 5m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a083917c1e0d496e98ec726859ded9e2 : FAILURE in 5m 09s","accounts_in_message":[],"_revision_number":5},{"id":"ccf6d34f3f89bb1c4e0cd81e57f95e8e7fea0e3d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2022-09-12 10:52:50.000000000","message":"Abandoned\n\nThis review is \u003e 12 weeks without comment, and failed testing the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a918a02c3f6821a3b98c6754b2dc2d9c54cd4e29","revisions":{"e259a01d3058f717c0ed50f686580da292ad2f9c":{"kind":"REWORK","_number":1,"created":"2021-04-21 21:54:45.000000000","uploader":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"ref":"refs/changes/48/787448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/48/787448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/1"}}},"commit":{"parents":[{"commit":"be273f7719c0158eca54e5151d144f01d89afa6d","subject":"Fix grammar in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/be273f7719c0158eca54e5151d144f01d89afa6d"}]}],"author":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-21 21:53:50.000000000","tz":-300},"committer":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-21 21:54:45.000000000","tz":-300},"subject":"Add NRPE check for errors in openvswitch status","message":"Add NRPE check for errors in openvswitch status\n\nPartial-bug: 1912847\nChange-Id: I62c66648898ba3fe59ec8523542b7a82a8b6c886\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/e259a01d3058f717c0ed50f686580da292ad2f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/e259a01d3058f717c0ed50f686580da292ad2f9c"}]},"branch":"refs/heads/master"},"b19d9254f61cd0070900d5b01e271ce3e1efa42b":{"kind":"REWORK","_number":2,"created":"2021-04-23 15:09:33.000000000","uploader":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"ref":"refs/changes/48/787448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/48/787448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/2"}}},"commit":{"parents":[{"commit":"be273f7719c0158eca54e5151d144f01d89afa6d","subject":"Fix grammar in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/be273f7719c0158eca54e5151d144f01d89afa6d"}]}],"author":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-21 21:53:50.000000000","tz":-300},"committer":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-23 15:08:48.000000000","tz":-300},"subject":"Add NRPE checks for openvswitch","message":"Add NRPE checks for openvswitch\n\nPartial-bug: 1912847\nChange-Id: I62c66648898ba3fe59ec8523542b7a82a8b6c886\nfunc-test-pr: https://github.com/openstack-charmers/zaza-openstack-tests/pull/566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/b19d9254f61cd0070900d5b01e271ce3e1efa42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/b19d9254f61cd0070900d5b01e271ce3e1efa42b"}]},"branch":"refs/heads/master"},"f1d4a3866d17956672b261ceed771746ecc0fb82":{"kind":"REWORK","_number":3,"created":"2021-04-23 19:05:31.000000000","uploader":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"ref":"refs/changes/48/787448/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/48/787448/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/3"}}},"commit":{"parents":[{"commit":"be273f7719c0158eca54e5151d144f01d89afa6d","subject":"Fix grammar in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/be273f7719c0158eca54e5151d144f01d89afa6d"}]}],"author":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-21 21:53:50.000000000","tz":-300},"committer":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-23 19:04:56.000000000","tz":-300},"subject":"Add NRPE checks for openvswitch","message":"Add NRPE checks for openvswitch\n\nPartial-bug: 1912847\nChange-Id: I62c66648898ba3fe59ec8523542b7a82a8b6c886\nfunc-test-pr: https://github.com/openstack-charmers/zaza-openstack-tests/pull/566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/f1d4a3866d17956672b261ceed771746ecc0fb82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/f1d4a3866d17956672b261ceed771746ecc0fb82"}]},"branch":"refs/heads/master"},"2b16732f4ffb36016cbf9cab1c4f92419a6c4f9e":{"kind":"REWORK","_number":4,"created":"2021-04-23 21:13:14.000000000","uploader":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"ref":"refs/changes/48/787448/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/48/787448/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/4"}}},"commit":{"parents":[{"commit":"be273f7719c0158eca54e5151d144f01d89afa6d","subject":"Fix grammar in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/be273f7719c0158eca54e5151d144f01d89afa6d"}]}],"author":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-21 21:53:50.000000000","tz":-300},"committer":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-23 21:12:08.000000000","tz":-300},"subject":"Add NRPE checks for openvswitch","message":"Add NRPE checks for openvswitch\n\nPartial-bug: 1912847\nChange-Id: I62c66648898ba3fe59ec8523542b7a82a8b6c886\nfunc-test-pr: https://github.com/openstack-charmers/zaza-openstack-tests/pull/566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/2b16732f4ffb36016cbf9cab1c4f92419a6c4f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/2b16732f4ffb36016cbf9cab1c4f92419a6c4f9e"}]},"branch":"refs/heads/master"},"a918a02c3f6821a3b98c6754b2dc2d9c54cd4e29":{"kind":"REWORK","_number":5,"created":"2021-04-26 16:10:05.000000000","uploader":{"_account_id":25011,"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","username":"afreiberger"},"ref":"refs/changes/48/787448/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/48/787448/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/48/787448/5"}}},"commit":{"parents":[{"commit":"be273f7719c0158eca54e5151d144f01d89afa6d","subject":"Fix grammar in README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/be273f7719c0158eca54e5151d144f01d89afa6d"}]}],"author":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-21 21:53:50.000000000","tz":-300},"committer":{"name":"Drew Freiberger","email":"drew.freiberger@canonical.com","date":"2021-04-26 16:09:57.000000000","tz":-300},"subject":"Add NRPE checks for openvswitch","message":"Add NRPE checks for openvswitch\n\nPartial-bug: 1912847\nChange-Id: I62c66648898ba3fe59ec8523542b7a82a8b6c886\nfunc-test-pr: https://github.com/openstack-charmers/zaza-openstack-tests/pull/566\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/a918a02c3f6821a3b98c6754b2dc2d9c54cd4e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/a918a02c3f6821a3b98c6754b2dc2d9c54cd4e29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
