)]}'
{"id":"openstack%2Fcharm-neutron-openvswitch~806677","triplet_id":"openstack%2Fcharm-neutron-openvswitch~master~I1546f45635bfd7af76001c1df4f99d219a9d8039","project":"openstack/charm-neutron-openvswitch","branch":"master","topic":"bug/1934129","hashtags":[],"change_id":"I1546f45635bfd7af76001c1df4f99d219a9d8039","subject":"Add extension fwaas_v2 based on neutron-api relation data","status":"MERGED","created":"2021-08-31 05:52:16.000000000","updated":"2021-09-13 12:24:27.000000000","submitted":"2021-09-13 12:24:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806677-bug/1934129","meta_rev_id":"621abad1d15f0567a5d7b451fa29850f252d4a41","_number":806677,"virtual_id_number":806677,"owner":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2021-09-12 04:28:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-13 12:24:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-13 12:14:52.000000000","_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-13 12:14:52.000000000","_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"value":0,"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"}],"reviewers":{"REVIEWER":[{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-31 06:29:49.000000000","updated_by":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"reviewer":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-31 06:47:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 08:27:36.000000000","updated_by":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"reviewer":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"state":"REVIEWER"},{"updated":"2021-09-13 12:14:52.000000000","updated_by":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"reviewer":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"state":"REVIEWER"}],"messages":[{"id":"97cc8ee3dcf2ef872f7d546730eeee92a184051b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-08-31 05:52:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d37e8cab5d026f6abd4b8b69ce4a9d2d6a6b4230","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-08-31 05:52:37.000000000","message":"Patch Set 1: Workflow-1\n\nSync charmhelpers required to get PR https://github.com/juju/charm-helpers/pull/635","accounts_in_message":[],"_revision_number":1},{"id":"3b054a84fe1a1a0720e530d17ad02ec10054126a","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-08-31 06:29:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/968/806677/1/check/osci-lint/9684cbc/index.html : SUCCESS in 4m 15s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/114/806677/1/check/tox-py35/11408b1/index.html : FAILURE in 7m 03s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/950/806677/1/check/tox-py36/950359a/index.html : FAILURE in 5m 50s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/816/806677/1/check/tox-py37/816d657/index.html : FAILURE in 7m 32s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/545/806677/1/check/tox-py38/545a8f7/index.html : FAILURE in 5m 41s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat migrate-ovn-focal-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"ac2f67840eb99d64230541c8a5948d195d0d40e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 06:47:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3c06b99ed6448b19f90b853123042f9 : SUCCESS in 8m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/384c499407d34e11a317683d2d23aa19 : FAILURE in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fbf080afbfe4a89a3da491ace68143a : FAILURE in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85993e696cc0482c810ffe5c386af160 : FAILURE in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c25df4491d48433786565f7651e324cc : FAILURE in 6m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f8d120467bb4059b7795f3325dfe357 : FAILURE in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"15fc4982135f91c1df7a8f0a49ee712ef21cd79d","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-08-31 09:33:24.000000000","message":"Patch Set 1:\n\nCI failures are expected without the changes from charmhelpers","accounts_in_message":[],"_revision_number":1},{"id":"d69422cd3fee518ca3e49fa438c41da34c517c42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-01 09:52:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42fb4de3f539b0b7cf7fcfc2f30df5ea9548d67e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 10:04:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f463ac9d809a40c8b84cad7b80f0bd92 : SUCCESS in 4m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ed28f5a3a4004435ac3f89c8f36d6c54 : SUCCESS in 10m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36dbc65a647a43219f4813b7c56c0f9d : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fcfb76bf54b4e62b937f72bcddeb49a : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ce00592672e4090b64df84db952c1a1 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e33df12b79e45d9ae7d896abe8f0a9b : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":2},{"id":"497aacfdc3c331591405b7b2f6179b400caf803b","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-09-01 11:24:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/75c/806677/2/check/osci-lint/75c40d5/index.html : SUCCESS in 4m 14s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/0fb/806677/2/check/tox-py35/0fbbe2a/index.html : SUCCESS in 5m 16s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/ede/806677/2/check/tox-py36/edef3cf/index.html : SUCCESS in 5m 52s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/558/806677/2/check/tox-py37/5584074/index.html : SUCCESS in 5m 15s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/bd6/806677/2/check/tox-py38/bd6f9f7/index.html : SUCCESS in 5m 46s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/d16/806677/2/check/migrate-ovn-focal-ussuri-dvr-snat/d16f3f7/index.html : FAILURE in 1h 25m 11s\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"f60e2e118d6efbb95354b173e66e94abdfcf58c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-07 13:41:55.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ca1901bffa4dd5ae65df0b044c59c1c701e4321e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 13:53:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0bd3b68ad747179c414dc30d39fc44 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f4577a8a4212478680c7c06766daa296 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b732daf6a18a4e32b71d89131e9f32a6 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bd16540e5df4de099fcbfebb5e54a65 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b785427111224cfcbeaaaf67b98a4460 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d14f4991ce314d2484c19aca02f9a502 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"787e541ae4c976f69ce94fdac0bd7a51c0441d55","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-09-07 15:16:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/3e6/806677/3/check/osci-lint/3e6e444/index.html : SUCCESS in 4m 36s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/dd6/806677/3/check/tox-py35/dd6435f/index.html : SUCCESS in 6m 12s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/684/806677/3/check/tox-py36/6845c61/index.html : SUCCESS in 6m 21s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/023/806677/3/check/tox-py37/02303ef/index.html : SUCCESS in 6m 32s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/025/806677/3/check/tox-py38/0259ea9/index.html : SUCCESS in 6m 48s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/9c7/806677/3/check/migrate-ovn-focal-ussuri-dvr-snat/9c7dca8/index.html : FAILURE in 1h 26m 53s\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":3},{"id":"e84f0f6ce645eddb0a480b79077977635ca76887","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-09-07 18:44:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca8730fbf84a43454c710bfeaccd0ec0b3aa88dc","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-08 06:20:11.000000000","message":"Patch Set 3:\n\nCI job migrate-ovn-focal-ussuri-dvr-snat failed with neutron-openvswitch units in \u0027waiting\u0027 state and status message \u0027Incomplete relations: neutron-plugin-api\u0027.\n\nunit-neutron-openvswitch juju logs shows the following message\n\"2021-09-07 14:18:00 INFO juju-log neutron-plugin-api relation\u0027s interface, neutron-plugin-api, is related awaiting the following data from the relationship: internal_host, internal_port.\"\n\nThese data are introduced in charm-neutron-api by https://review.opendev.org/c/openstack/charm-neutron-api/+/806676 but the neutron-plugin-api relation data is not updated.\n\nSubmitted a patch for the same -\nhttps://review.opendev.org/c/openstack/charm-neutron-api/+/807822\n\nThe CI build need to be retriggered once 807822 is merged.","accounts_in_message":[],"_revision_number":3},{"id":"e824f13c291a6ede092c662f8a65a2a65d536143","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-09 09:05:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3cfe14d5b983613a8c5f0924bc38a065871f3fc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 09:12:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6624b3407b464d0d9e882ea505568b17 : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ebbb0776170e406a9259d9db6af90632 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6a12001eb1f41808a326b77ee53340a : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e87d4c1649d74fce87fb2434d530cf2c : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6579735b235546d5b9ad0884120b3761 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99ce83babc914f4db81828dcf840787b : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":4},{"id":"6fb2497d562aa4aa9546b80a1e0060a3671b0eb7","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-09 09:20:09.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nHi Ed,\n\nPlease see my comments inline and review again.","accounts_in_message":[],"_revision_number":4},{"id":"4519d4172ba20e69303207c48be76ea59a39e155","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-09-09 10:40:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild FAILURE by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/1d0/806677/4/check/osci-lint/1d080ba/index.html : SUCCESS in 4m 47s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/77b/806677/4/check/tox-py35/77bbfd5/index.html : SUCCESS in 6m 11s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/b6a/806677/4/check/tox-py36/b6a3d77/index.html : SUCCESS in 6m 54s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/9dd/806677/4/check/tox-py37/9dd139f/index.html : SUCCESS in 6m 44s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/49e/806677/4/check/tox-py38/49e821d/index.html : SUCCESS in 6m 36s\n- xenial-mitaka_ovs xenial-mitaka_ovs : SKIPPED\n- bionic-queens-dvr-snat bionic-queens-dvr-snat : SKIPPED\n- bionic-stein-dvr-snat bionic-stein-dvr-snat : SKIPPED\n- bionic-train-dvr-snat bionic-train-dvr-snat : SKIPPED\n- bionic-ussuri-dvr-snat bionic-ussuri-dvr-snat : SKIPPED\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/92b/806677/4/check/migrate-ovn-focal-ussuri-dvr-snat/92be4f0/index.html : FAILURE in 1h 26m 29s\n- migrate-ovn-focal-victoria-dvr-snat migrate-ovn-focal-victoria-dvr-snat : SKIPPED","accounts_in_message":[],"_revision_number":4},{"id":"0e399e03f28ba2c2ff99b18bad3e36b5a71594ca","author":{"_account_id":6737,"name":"Edward Hope-Morley","email":"edward.hope-morley@canonical.com","username":"hopem"},"date":"2021-09-10 11:09:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3c4c575df172adfa04cf8a5b23a98c61df7d32c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"date":"2021-09-12 01:36:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"75da11e58063828b56ba51e9424fc9d811c97aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-12 01:50:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a62ac04eda1941868429ce5cf5edb9ca : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/62743077a7ca489c8af3c08ee63fc460 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08aa091e7c014df3a904e5287309a0ac : SUCCESS in 10m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46e321df40db44e4a06b469d7e749cbd : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeed6b8b9b3f460a96caef32c326ed15 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ca39ae5a0bd4308bbd66a2fd24dcdb8 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":5},{"id":"63c0dd47517de0b2afdc207a039a59910dcf186e","author":{"_account_id":20648,"name":"Canonical CI","email":"uosci-testing-bot@ubuntu.com","username":"uosci-testing-bot","tags":["SERVICE_USER"]},"date":"2021-09-12 04:28:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild SUCCESS by Canonical OpenStack Charm CI.\n\nRe-test comment options:\n\n- \u0027recheck\u0027 (run/re-run the default upstream and charm test sets)\n- \u0027charm-recheck\u0027 (run/re-run the default charm verify test set) with Jenkins and Zuul\n- \u0027zosci-recheck\u0027 (run/re-run the default charm verify test set) with only Zuul\n\nFor more info, see http://docs.openstack.org/developer/charm-guide/ and https://wiki.openstack.org/wiki/ThirdPartySystems/Canonical_Charm_CI.\n\n\n- osci-lint https://openstack-ci-reports.ubuntu.com/artifacts/bc5/806677/5/check/osci-lint/bc5954c/index.html : SUCCESS in 3m 37s\n- tox-py35 https://openstack-ci-reports.ubuntu.com/artifacts/ff0/806677/5/check/tox-py35/ff0976a/index.html : SUCCESS in 5m 17s\n- tox-py36 https://openstack-ci-reports.ubuntu.com/artifacts/b58/806677/5/check/tox-py36/b580297/index.html : SUCCESS in 5m 24s\n- tox-py37 https://openstack-ci-reports.ubuntu.com/artifacts/911/806677/5/check/tox-py37/91122cd/index.html : SUCCESS in 5m 05s\n- tox-py38 https://openstack-ci-reports.ubuntu.com/artifacts/9a5/806677/5/check/tox-py38/9a55959/index.html : SUCCESS in 5m 24s\n- xenial-mitaka_ovs https://openstack-ci-reports.ubuntu.com/artifacts/ff7/806677/5/check/xenial-mitaka_ovs/ff75a42/index.html : SUCCESS in 56m 56s\n- bionic-queens-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/918/806677/5/check/bionic-queens-dvr-snat/918e246/index.html : SUCCESS in 1h 19m 32s\n- bionic-stein-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/bf8/806677/5/check/bionic-stein-dvr-snat/bf884b9/index.html : SUCCESS in 1h 24m 16s\n- bionic-train-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/f48/806677/5/check/bionic-train-dvr-snat/f48ba61/index.html : SUCCESS in 1h 23m 46s\n- bionic-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/efa/806677/5/check/bionic-ussuri-dvr-snat/efac524/index.html : SUCCESS in 1h 31m 52s\n- migrate-ovn-focal-ussuri-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/a5a/806677/5/check/migrate-ovn-focal-ussuri-dvr-snat/a5a40d2/index.html : SUCCESS in 1h 13m 19s\n- migrate-ovn-focal-victoria-dvr-snat https://openstack-ci-reports.ubuntu.com/artifacts/4fd/806677/5/check/migrate-ovn-focal-victoria-dvr-snat/4fd0485/index.html : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":5},{"id":"b68c9e177b204bae4e17daf9bab192e58e4f4d96","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2021-09-13 12:14:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0db8ef78ede5cd5426012dad5afd707212f0c3ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 12:15:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"22d59e66cd24f54544ca0256249762bb6731150d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 12:24:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7404a873af94ca7983f7ed4e7b451ae : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ce56e03689e54f55a3cb08b21ee10ff8 : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c4c3780aa4d47fc9eae4a6a40fccd2d : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/487975d3b0314194b667ebd1ee3af62f : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":5},{"id":"621abad1d15f0567a5d7b451fa29850f252d4a41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 12:24:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d3daa09bc49bcc5c66791565e1720a9a484ea149","revisions":{"de622e87588c09032ddea6763cafaf3c37b23b90":{"kind":"REWORK","_number":1,"created":"2021-08-31 05:52:16.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/77/806677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/77/806677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/1"}}},"commit":{"parents":[{"commit":"13b024a3a2cd90ddfc9ccad1e3daebc421eb40af","subject":"Merge \"Replace reference to \u0027juju set\u0027 with \u0027juju config\u0027.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/13b024a3a2cd90ddfc9ccad1e3daebc421eb40af"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-31 05:44:43.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-31 05:52:13.000000000","tz":330},"subject":"Add extension fwaas_v2 based on neutron-api relation data","message":"Add extension fwaas_v2 based on neutron-api relation data\n\nfwaas_v2 extension is added in l3_agent.ini by default from stein.\nHowever with the newly introduced neutron-api configuration parameter\nenable-fwaas, the extension have to added only when it is set to true\non neutron-api charm.\n\nneutron-api charm adds enabled-fwaas as relation data in the commit\n53013249b205abd1ae51de22f0504c98b5ac08df.\nThis patch adds fwaas_v2 to l3_extension_plugins when enable-fwaas\nis set to true in relation data.\nThis patch also purges python3-neutron-fwaas from victoria release.\n\nTODO: Sync charmhelpers to get related changes\nhttps://github.com/juju/charm-helpers/pull/635\n\nCloses-Bug: #1934129\nChange-Id: I1546f45635bfd7af76001c1df4f99d219a9d8039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/de622e87588c09032ddea6763cafaf3c37b23b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/de622e87588c09032ddea6763cafaf3c37b23b90"}]},"branch":"refs/heads/master"},"e00a3003261228edeb2023b23339e6068c033b7f":{"kind":"REWORK","_number":2,"created":"2021-09-01 09:52:43.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/77/806677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/77/806677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/2"}}},"commit":{"parents":[{"commit":"13b024a3a2cd90ddfc9ccad1e3daebc421eb40af","subject":"Merge \"Replace reference to \u0027juju set\u0027 with \u0027juju config\u0027.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/13b024a3a2cd90ddfc9ccad1e3daebc421eb40af"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-31 05:44:43.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-09-01 09:52:07.000000000","tz":330},"subject":"Add extension fwaas_v2 based on neutron-api relation data","message":"Add extension fwaas_v2 based on neutron-api relation data\n\nfwaas_v2 extension is added in l3_agent.ini by default from stein.\nHowever with the newly introduced neutron-api configuration parameter\nenable-fwaas, the extension have to added only when it is set to true\non neutron-api charm.\n\nneutron-api charm adds enabled-fwaas as relation data in the commit\n53013249b205abd1ae51de22f0504c98b5ac08df.\nThis patch adds fwaas_v2 to l3_extension_plugins when enable-fwaas\nis set to true in relation data.\nThis patch also purges python3-neutron-fwaas from victoria release.\n\nSynced charmhelpers to get related changes\nhttps://github.com/juju/charm-helpers/pull/635\n\nCloses-Bug: #1934129\nChange-Id: I1546f45635bfd7af76001c1df4f99d219a9d8039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/e00a3003261228edeb2023b23339e6068c033b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/e00a3003261228edeb2023b23339e6068c033b7f"}]},"branch":"refs/heads/master"},"f58c748071b845d6ae19271c1773075a12fd4159":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-07 13:41:55.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/77/806677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/77/806677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/3"}}},"commit":{"parents":[{"commit":"13b024a3a2cd90ddfc9ccad1e3daebc421eb40af","subject":"Merge \"Replace reference to \u0027juju set\u0027 with \u0027juju config\u0027.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/13b024a3a2cd90ddfc9ccad1e3daebc421eb40af"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-31 05:44:43.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-09-07 13:40:34.000000000","tz":330},"subject":"Add extension fwaas_v2 based on neutron-api relation data","message":"Add extension fwaas_v2 based on neutron-api relation data\n\nfwaas_v2 extension is added in l3_agent.ini by default from stein.\nHowever with the newly introduced neutron-api configuration parameter\nenable-fwaas, the extension have to added only when it is set to true\non neutron-api charm.\n\nneutron-api charm adds enabled-fwaas as relation data in the commit\nhttps://review.opendev.org/c/openstack/charm-neutron-api/+/806676.\nThis patch adds fwaas_v2 to l3_extension_plugins when enable-fwaas\nis set to true in relation data.\nThis patch also purges python3-neutron-fwaas from victoria release.\n\nSynced charmhelpers to get related changes\nhttps://github.com/juju/charm-helpers/pull/635\n\nCloses-Bug: #1934129\nChange-Id: I1546f45635bfd7af76001c1df4f99d219a9d8039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/f58c748071b845d6ae19271c1773075a12fd4159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/f58c748071b845d6ae19271c1773075a12fd4159"}]},"branch":"refs/heads/master"},"89724d804915f68b6e391cdfe9f007b08bdae997":{"kind":"REWORK","_number":4,"created":"2021-09-09 09:05:48.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/77/806677/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/77/806677/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/4"}}},"commit":{"parents":[{"commit":"13b024a3a2cd90ddfc9ccad1e3daebc421eb40af","subject":"Merge \"Replace reference to \u0027juju set\u0027 with \u0027juju config\u0027.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/13b024a3a2cd90ddfc9ccad1e3daebc421eb40af"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-31 05:44:43.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-09-09 09:02:41.000000000","tz":330},"subject":"Add extension fwaas_v2 based on neutron-api relation data","message":"Add extension fwaas_v2 based on neutron-api relation data\n\nfwaas_v2 extension is added in l3_agent.ini by default from stein.\nHowever with the newly introduced neutron-api configuration parameter\nenable-fwaas, the extension have to added only when it is set to true\non neutron-api charm.\n\nneutron-api charm adds enabled-fwaas as relation data in the commit\nhttps://review.opendev.org/c/openstack/charm-neutron-api/+/806676.\nThis patch adds fwaas_v2 to l3_extension_plugins when enable-fwaas\nis set to true in relation data.\n\nSynced charmhelpers to get related changes\nhttps://github.com/juju/charm-helpers/pull/635\n\nCloses-Bug: #1934129\nChange-Id: I1546f45635bfd7af76001c1df4f99d219a9d8039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/89724d804915f68b6e391cdfe9f007b08bdae997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/89724d804915f68b6e391cdfe9f007b08bdae997"}]},"branch":"refs/heads/master"},"d3daa09bc49bcc5c66791565e1720a9a484ea149":{"kind":"REWORK","_number":5,"created":"2021-09-12 01:36:37.000000000","uploader":{"_account_id":10366,"name":"Hemanth N","email":"hemanth.nakkina@canonical.com","username":"Hemanth"},"ref":"refs/changes/77/806677/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/charm-neutron-openvswitch","ref":"refs/changes/77/806677/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/charm-neutron-openvswitch refs/changes/77/806677/5"}}},"commit":{"parents":[{"commit":"13b024a3a2cd90ddfc9ccad1e3daebc421eb40af","subject":"Merge \"Replace reference to \u0027juju set\u0027 with \u0027juju config\u0027.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/13b024a3a2cd90ddfc9ccad1e3daebc421eb40af"}]}],"author":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-08-31 05:44:43.000000000","tz":330},"committer":{"name":"Hemanth Nakkina","email":"hemanth.nakkina@canonical.com","date":"2021-09-12 01:35:11.000000000","tz":330},"subject":"Add extension fwaas_v2 based on neutron-api relation data","message":"Add extension fwaas_v2 based on neutron-api relation data\n\nfwaas_v2 extension is added in l3_agent.ini by default from stein.\nHowever with the newly introduced neutron-api configuration parameter\nenable-fwaas, the extension have to added only when it is set to true\non neutron-api charm.\n\nneutron-api charm adds enabled-fwaas as relation data in the commit\nhttps://review.opendev.org/c/openstack/charm-neutron-api/+/806676.\nThis patch removes special handling of fwaas_v2 as l3_extension_plugins\nis already updated in relation data based on neutron-api enable-fwaas\nconfiguration.\n\nAdd python3-neutron-fwaas in py3 package list only for rocky release. The\npackage is already part of dependency of neutron-l3-agent until Ussuri\nrelease and the dependency will be removed from Victoria release in\npackage control files.\n\nSynced charmhelpers to get related changes\nhttps://github.com/juju/charm-helpers/pull/635\n\nCloses-Bug: #1934129\nChange-Id: I1546f45635bfd7af76001c1df4f99d219a9d8039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/d3daa09bc49bcc5c66791565e1720a9a484ea149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/charm-neutron-openvswitch/commit/d3daa09bc49bcc5c66791565e1720a9a484ea149"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
